)]}'
{"id":"openstack%2Fmonasca-notification~852059","triplet_id":"openstack%2Fmonasca-notification~master~Ifa7489b71e48a8c25ae0e00e26aa3d2f1c5e15b0","project":"openstack/monasca-notification","branch":"master","attention_set":{},"removed_from_attention_set":{"28743":{"account":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"last_update":"2026-02-03 22:35:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ifa7489b71e48a8c25ae0e00e26aa3d2f1c5e15b0","subject":"Replace abc.abstractproperty with property and abc.abstractmethod","status":"ABANDONED","created":"2022-08-04 01:05:12.000000000","updated":"2026-02-03 22:35:57.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"456db6d90a40c010ef7a8b0392ca0e7192ac8475","_number":852059,"virtual_id_number":852059,"owner":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-04 01:54:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-04 01:54:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-03 22:35:57.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"7d8e6d019619fb4d877c9480484082e3ffe568a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2022-08-04 01:05:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f082ca5693d85bcb7133bd6712df085332061ca5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 01:54:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/059705cbee154cf3b2d8e6dd515ade9c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f60fdfeb6cee4e5a8b699dfce6d3c747 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8480f09d5c346f3bebc066c328ef184 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d9bdf967a094ba88045c8faddbaa2b6 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72d8c5fcc08945fda7b066df91870d5b : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c781fa7c250b4bc08a9ddfde61f8c6e2 : SUCCESS in 5m 14s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/cd674c4a40c3441bbe130e695ab37af3 : FAILURE in 5m 35s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/ce488e28afca4ca18935c953e1b7f4e2 : SUCCESS in 47m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da5897efd667d9719f7e6dcd3ab415da26a98a8d","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2022-08-04 04:01:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67d87f89a696ca3c592c04b97d17c81a5e496930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 04:57:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab008515fd6144aaa216bca4134592c8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/739d6b89c6b94ce8a4e35ac842c7a97a : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1742ba4943704a7b92f1bace5c5a7ee9 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dac81bd5e2b7419cbe52fc5c8517e09b : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9199daa2fb2741cba6a9c6b339d9ed21 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59ff39186e7a47afb0620f895b46e128 : SUCCESS in 4m 17s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/71e01f8655cd424fbf48c1125830ddb0 : FAILURE in 5m 59s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/b2652f1e6c254910bce1ab58dd236f44 : SUCCESS in 50m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d7a7914991f74f9137d2f1b72251a61b7d31502","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2022-08-05 01:15:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fd16b37f993a4c6c3422e20bc959e6318e77f045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 01:54:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5a9da9c28e14865a49df0276688cc77\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36ab60e8aa844bfe9a6dc9a0f18edf3a : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb0e23b633074675b1d11e7ebfa5f80c : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58983b9d47e34e319dbf4b7166eee901 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5838c1d0c48e42b88190edef6c84d93d : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7484ed062d92420480cef2ed4daad725 : SUCCESS in 3m 50s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/891867d1b1854def9f206e490990fa23 : FAILURE in 5m 02s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/cf2e95714b6846809f12272ec0f3cb11 : FAILURE in 37m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"312caf6e4a3a7308c72a9e7abac6ca929bb7c1fd","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2022-08-10 08:52:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1cb77a92d343dc36eb5f326344b78f10447fb96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 09:37:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fa2405d49b248728a6186c3bce04782\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/132bc4f803de416693bbae1a92d04a34 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c743245aa5d4d64a079fa0180515669 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c4d872ec7474c6d9e5e476d2f79ba25 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2b6d9ae6c9e40dcaed6eaeb1c218e59 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ca5347727b84823bb7752e56c88f813 : SUCCESS in 4m 12s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/e0e5040cb6d04de593f1783c7ed252f1 : FAILURE in 4m 56s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/0a5e8105b8bc45938ff043f06409ebda : SUCCESS in 42m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"456db6d90a40c010ef7a8b0392ca0e7192ac8475","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-03 22:35:57.000000000","message":"Abandoned\n\nThis project has been retired and no longer accepts code changes via https://review.opendev.org. New changes can be uploaded if the project is restored.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dfceb942049c6277b36957707b6dd7bfefb2c1cf","revisions":{"dfceb942049c6277b36957707b6dd7bfefb2c1cf":{"kind":"REWORK","_number":1,"created":"2022-08-04 01:05:12.000000000","uploader":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"ref":"refs/changes/59/852059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/59/852059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/59/852059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/59/852059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/59/852059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/59/852059/1"}}},"commit":{"parents":[{"commit":"5225304a268fbd7e5edc30ab6d2b0cbf3b92b565","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/5225304a268fbd7e5edc30ab6d2b0cbf3b92b565"}]}],"author":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2022-08-04 01:04:35.000000000","tz":480},"committer":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2022-08-04 01:04:35.000000000","tz":480},"subject":"Replace abc.abstractproperty with property and abc.abstractmethod","message":"Replace abc.abstractproperty with property and abc.abstractmethod\n\nReplace abc.abstractproperty with property and abc.abstractmethod,\nas abc.abstractproperty has been deprecated since python3.3[1]\n\n[1]https://docs.python.org/3.8/whatsnew/3.3.html?highlight\u003ddeprecated#abc\n\nChange-Id: Ifa7489b71e48a8c25ae0e00e26aa3d2f1c5e15b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/dfceb942049c6277b36957707b6dd7bfefb2c1cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/dfceb942049c6277b36957707b6dd7bfefb2c1cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
