)]}'
{"id":"openstack%2Fmonasca-notification~894972","triplet_id":"openstack%2Fmonasca-notification~master~I918a8bbe64739732b84e2dc33d984454c0aa9aae","project":"openstack/monasca-notification","branch":"master","topic":"release-health-check-cwr-bobcat","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-02-14 16:56:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I918a8bbe64739732b84e2dc33d984454c0aa9aae","subject":"DNM: gate health test","status":"ABANDONED","created":"2023-09-13 19:20:07.000000000","updated":"2024-02-14 16:56:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d8360a7ae2764b19c42f85bfc1fdf461f8c98f6b","_number":894972,"virtual_id_number":894972,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-13 19:47:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-13 19:47:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aaa55f13b646dbf5f6e6f7c368b2394dcd5eaf6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-13 19:20:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d2d67e6877d083d3ccdc01ca96d6edade7d835a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 19:47:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/012c625642024929bc216b0358b28119\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b2c83e4875f4028853d886f4c2c962e : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6708358442cc42b0a61fd31ead35543a : FAILURE in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dfe8c6caf954a18a2fd989e9c0d891d : FAILURE in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1da0486a98c644558ba898e3e7aac697 : FAILURE in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/518255c615e540899034e56a2df9e7c7 : FAILURE in 2m 19s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/34cb62c2023f40138a94eecadaae6d8a : FAILURE in 5m 41s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/3f5c5f75b4f3418cb4445df4d22ca165 : FAILURE in 25m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8360a7ae2764b19c42f85bfc1fdf461f8c98f6b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-14 16:56:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e591d484345de5f18cd5eb8dc9d43b2825d5f685","revisions":{"e591d484345de5f18cd5eb8dc9d43b2825d5f685":{"kind":"REWORK","_number":1,"created":"2023-09-13 19:20:07.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/72/894972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/72/894972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/72/894972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/72/894972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/72/894972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/72/894972/1"}}},"commit":{"parents":[{"commit":"5225304a268fbd7e5edc30ab6d2b0cbf3b92b565","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/5225304a268fbd7e5edc30ab6d2b0cbf3b92b565"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-13 19:19:58.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-13 19:20:05.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: I918a8bbe64739732b84e2dc33d984454c0aa9aae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/e591d484345de5f18cd5eb8dc9d43b2825d5f685"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/e591d484345de5f18cd5eb8dc9d43b2825d5f685"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
