)]}'
{"id":"openstack%2Fmonasca-notification~895382","triplet_id":"openstack%2Fmonasca-notification~master~I9b8bb5011d13e84dafc49dbdaba794d600ce6815","project":"openstack/monasca-notification","branch":"master","topic":"reno-2023.2","hashtags":[],"change_id":"I9b8bb5011d13e84dafc49dbdaba794d600ce6815","subject":"Update master for stable/2023.2","status":"ABANDONED","created":"2023-09-15 14:08:40.000000000","updated":"2026-02-03 22:35:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"38d575b291b8e376574e52f13a0a0d8f3530d661","_number":895382,"virtual_id_number":895382,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-07 09:35:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-15 14:24:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-03 22:35:04.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"37f174f2b8fb7e21fdb2a31f1308dd8f85bc3d3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-09-15 14:08:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"155ae1615b15c6d8183ff213c5859a15ca84649f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 14:24:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3413a6588c764c7483552c9e8b31cb09\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e538496bdfbf45e495dc651fc77e622f : FAILURE in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dda3110ed8994346bc2d79408b07422e : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3a87e9f39aa593c672fba7daa9de667a487254e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-03-07 09:22:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_22816\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"b32f47330a6fab9e2ad42385ca53b1d15be49c01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 09:35:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8416137947c445cacdb9227f5a64505\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b6d67095c6141349e593213d2007c14 : SUCCESS in 2m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0c79df501664e4e99fcd6b1b2bcc38a : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":2},{"id":"38d575b291b8e376574e52f13a0a0d8f3530d661","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-03 22:35:04.000000000","message":"Abandoned\n\nThis project has been retired and no longer accepts code changes via https://review.opendev.org. New changes can be uploaded if the project is restored.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9b3972862408a4b383a5e334f376a54adaa48976","revisions":{"17e099ae6bf68af2dc6785e82c6de931328b442f":{"kind":"REWORK","_number":1,"created":"2023-09-15 14:08:40.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/82/895382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/82/895382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/82/895382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/82/895382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/82/895382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/82/895382/1"}}},"commit":{"parents":[{"commit":"5225304a268fbd7e5edc30ab6d2b0cbf3b92b565","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/5225304a268fbd7e5edc30ab6d2b0cbf3b92b565"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-15 14:08:40.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-15 14:08:40.000000000","tz":0},"subject":"Update master for stable/2023.2","message":"Update master for stable/2023.2\n\nAdd file to the reno documentation build to show release notes for\nstable/2023.2.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/2023.2.\n\nSem-Ver: feature\nChange-Id: I9b8bb5011d13e84dafc49dbdaba794d600ce6815\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/17e099ae6bf68af2dc6785e82c6de931328b442f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/17e099ae6bf68af2dc6785e82c6de931328b442f"}]},"branch":"refs/heads/master"},"9b3972862408a4b383a5e334f376a54adaa48976":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-07 09:22:47.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"ref":"refs/changes/82/895382/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/82/895382/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/82/895382/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/82/895382/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/82/895382/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/82/895382/2"}}},"commit":{"parents":[{"commit":"13121a1059492dc36a885e443a728b53cfafe3d8","subject":"Making the gates happy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/13121a1059492dc36a885e443a728b53cfafe3d8"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-15 14:08:40.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-07 09:22:47.000000000","tz":0},"subject":"Update master for stable/2023.2","message":"Update master for stable/2023.2\n\nAdd file to the reno documentation build to show release notes for\nstable/2023.2.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/2023.2.\n\nSem-Ver: feature\nChange-Id: I9b8bb5011d13e84dafc49dbdaba794d600ce6815\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/9b3972862408a4b383a5e334f376a54adaa48976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/9b3972862408a4b383a5e334f376a54adaa48976"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
