)]}'
{"id":"openstack%2Fmonasca-notification~911410","triplet_id":"openstack%2Fmonasca-notification~master~Iefc7fdc89a398623aa3515c89240b53a49c10282","project":"openstack/monasca-notification","branch":"master","attention_set":{},"removed_from_attention_set":{"35988":{"account":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"last_update":"2024-03-07 08:36:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iefc7fdc89a398623aa3515c89240b53a49c10282","subject":"Making the gates happy","status":"MERGED","created":"2024-03-06 08:07:18.000000000","updated":"2024-03-07 08:37:26.000000000","submitted":"2024-03-07 08:36:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911410","meta_rev_id":"8026498a7af43e1a2e0540b8f55ceba75dc653ec","_number":911410,"virtual_id_number":911410,"owner":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-07 08:36:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"all":[{"value":2,"date":"2024-03-07 08:23:18.000000000","_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"all":[{"value":1,"date":"2024-03-07 08:23:43.000000000","_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-06 08:41:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-07 08:23:18.000000000","updated_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"REVIEWER"}],"messages":[{"id":"9702a64e604cb47244ebb3c83ce8c0fff7b1c188","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-03-06 08:07:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0424dd5981e370f1018f1472c2231aeff4195a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 08:41:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/010c60fd58834489947d63617a4ae59b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cf5b95e4b484b249a3b651c457ebee7 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d666049ef8d4b03acc184dc1c1a85ee : FAILURE in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84003da4d77f429b92d6437e079e8574 : FAILURE in 2m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/818906510c3d49758d38d0b33575e1dc : FAILURE in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/246c005c891b4a2c86fb1d4bbce8eb11 : FAILURE in 2m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39195542ffd542dd94948be6dbec93ef : SUCCESS in 2m 19s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/05ee678b03ba4a29a9460f8324df8763 : FAILURE in 5m 48s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/6741ca101f6a4a919cb700b2d8053315 : FAILURE in 32m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab1ea0d133d9765ad7408ad48f8f513cf70ec34c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-03-06 11:02:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d23af9078462f27e632899fa1d4dff1520b22cd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-03-06 11:03:59.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3f0e884fc1affe236a2af78f0aecb098daaabe57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-03-06 11:07:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"eae132ce7d7ffd94c990eb82939b63a498ce239b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-03-06 11:14:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"99304ef266e3e0828b55215adb5a8e2457a32787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 12:10:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/64ce42b76c9b4357b886a10e3fea103c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c82276b54cb94b239c65a08b0238c80a : SUCCESS in 1m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8291ff4118340df8ff30f99aec74583 : FAILURE in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f1fe3bc48824d8d88db9f17704b1eaf : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41fb77bca36040f3b3ae4e58cdfade74 : FAILURE in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83b57ffa278949bba143708eab7fadd5 : FAILURE in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f342e3fccced4d19b99133439d330332 : FAILURE in 3m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd03b9f13ab0445095aea5e00079673d : SUCCESS in 3m 26s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/d55e74aab94c4ac694e820de363a1ded : SUCCESS in 6m 04s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/bdf28ec38afb43ff9fde5961effdaebb : FAILURE in 47m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fbd30e56abe98c06d0ea7322b125794f6c95fd3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-03-06 13:40:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"05c736f7174c6384cb91bbb7959ab0fbaa1d3db3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 14:33:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87c6663fae414bf2859c32be5d152969\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/473944aca38b495c8579e020600fe59a : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51a6472310334b3eac1b302c276ef2d5 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faa4f022753144ee9d3409c98ed427aa : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc4d9d3b407b48c5860393a43a72c53e : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31919de65b56473987f07816f173ca74 : SUCCESS in 4m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51939ddd18b6400bb383968a59f0b5ea : SUCCESS in 2m 37s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/32b47f2d4a454e9ea3f8b90badbf791f : SUCCESS in 6m 14s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/a4d510ebdaf6481ea51bf5fe1d141952 : FAILURE in 46m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"76f616fb7ff8ed8b958ab820a3474ac557d1b40a","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-03-07 08:23:18.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"68aa4b59c4d0f99e892918a5fc94104ba1a79ea9","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-03-07 08:23:43.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7f60ef117ee51779bfadd12427775fc58c3f0311","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 08:24:02.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"dd22021e23ba2967bcaa7e975ff9862bc8ebf3ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 08:36:27.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0157b8c187d42478b6198e7e07060ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05110a1550f2479b9996b0941baa5db3 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6146e436335483bbc38eb6de51b1ec7 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd15be7ad37d45c885e73821f33ad321 : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d20885c1aac49f1961b678f3ba16ff3 : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":6},{"id":"4e70a5697c981bf4b93cc8b79ab91610599c672c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 08:36:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"8026498a7af43e1a2e0540b8f55ceba75dc653ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 08:37:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58846e6cdea04177a8ab2bd95f845666\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d67f852830d74ab2b33ededf3a6ab713 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"13121a1059492dc36a885e443a728b53cfafe3d8","revisions":{"d7af0c68a8e2afa0c1614a7411e87cd0999d5c4c":{"kind":"REWORK","_number":1,"created":"2024-03-06 08:07:18.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/10/911410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/10/911410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/1"}}},"commit":{"parents":[{"commit":"5225304a268fbd7e5edc30ab6d2b0cbf3b92b565","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/5225304a268fbd7e5edc30ab6d2b0cbf3b92b565"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-03-06 08:08:56.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-03-06 08:09:25.000000000","tz":0},"subject":"fix: allowlist_externals added to tox","message":"fix: allowlist_externals added to tox\n\nChange-Id: Iefc7fdc89a398623aa3515c89240b53a49c10282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/d7af0c68a8e2afa0c1614a7411e87cd0999d5c4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/d7af0c68a8e2afa0c1614a7411e87cd0999d5c4c"}]},"branch":"refs/heads/master"},"4d88bc104f482510d086a2c0b1dfe3f7ef35cdc2":{"kind":"REWORK","_number":2,"created":"2024-03-06 11:02:52.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/10/911410/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/10/911410/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/2"}}},"commit":{"parents":[{"commit":"5225304a268fbd7e5edc30ab6d2b0cbf3b92b565","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/5225304a268fbd7e5edc30ab6d2b0cbf3b92b565"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-03-06 08:08:56.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-03-06 10:58:29.000000000","tz":0},"subject":"Making happy gates","message":"Making happy gates\n\nfix: allowlist_externals added to tox\n\nfix: upgrading gevent\n\nfix: pep8 issue fixed for slack_notifier\n\nfix: change env list\n\nChange-Id: Iefc7fdc89a398623aa3515c89240b53a49c10282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/4d88bc104f482510d086a2c0b1dfe3f7ef35cdc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/4d88bc104f482510d086a2c0b1dfe3f7ef35cdc2"}]},"branch":"refs/heads/master"},"5a9b99c9a0d21434c6ecbab5035d3e06de8eaee5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-03-06 11:03:59.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/10/911410/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/10/911410/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/3"}}},"commit":{"parents":[{"commit":"5225304a268fbd7e5edc30ab6d2b0cbf3b92b565","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/5225304a268fbd7e5edc30ab6d2b0cbf3b92b565"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-03-06 08:08:56.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-03-06 11:05:48.000000000","tz":0},"subject":"Making the gates happy","message":"Making the gates happy\n\nfix: allowlist_externals added to tox\n\nfix: upgrading gevent\n\nfix: pep8 issue fixed for slack_notifier\n\nfix: change env list\n\nChange-Id: Iefc7fdc89a398623aa3515c89240b53a49c10282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/5a9b99c9a0d21434c6ecbab5035d3e06de8eaee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/5a9b99c9a0d21434c6ecbab5035d3e06de8eaee5"}]},"branch":"refs/heads/master"},"50bf844e4f78458e3db68fd1b81dffcfbebab3aa":{"kind":"REWORK","_number":4,"created":"2024-03-06 11:07:39.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/10/911410/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/10/911410/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/4"}}},"commit":{"parents":[{"commit":"5225304a268fbd7e5edc30ab6d2b0cbf3b92b565","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/5225304a268fbd7e5edc30ab6d2b0cbf3b92b565"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-03-06 08:08:56.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-03-06 11:09:41.000000000","tz":0},"subject":"Making the gates happy","message":"Making the gates happy\n\nfix: allowlist_externals added to tox\n\nfix: upgrading gevent\n\nfix: pep8 issue fixed for slack_notifier\n\nfix: change env list\n\nChange-Id: Iefc7fdc89a398623aa3515c89240b53a49c10282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/50bf844e4f78458e3db68fd1b81dffcfbebab3aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/50bf844e4f78458e3db68fd1b81dffcfbebab3aa"}]},"branch":"refs/heads/master"},"34dd31d5e16ea24f8cad8c3cf244661cf828f70a":{"kind":"REWORK","_number":5,"created":"2024-03-06 11:14:12.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/10/911410/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/10/911410/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/5"}}},"commit":{"parents":[{"commit":"5225304a268fbd7e5edc30ab6d2b0cbf3b92b565","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/5225304a268fbd7e5edc30ab6d2b0cbf3b92b565"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-03-06 08:08:56.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-03-06 11:16:14.000000000","tz":0},"subject":"Making the gates happy","message":"Making the gates happy\n\nfix: allowlist_externals added to tox\n\nfix: upgrading gevent\n\nfix: pep8 issue fixed for slack_notifier\n\nfix: change env list\n\nChange-Id: Iefc7fdc89a398623aa3515c89240b53a49c10282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/34dd31d5e16ea24f8cad8c3cf244661cf828f70a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/34dd31d5e16ea24f8cad8c3cf244661cf828f70a"}]},"branch":"refs/heads/master"},"13121a1059492dc36a885e443a728b53cfafe3d8":{"kind":"REWORK","_number":6,"created":"2024-03-06 13:40:41.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/10/911410/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/10/911410/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/10/911410/6"}}},"commit":{"parents":[{"commit":"5225304a268fbd7e5edc30ab6d2b0cbf3b92b565","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/5225304a268fbd7e5edc30ab6d2b0cbf3b92b565"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-03-06 08:08:56.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-03-06 13:42:44.000000000","tz":0},"subject":"Making the gates happy","message":"Making the gates happy\n\nfix: allowlist_externals added to tox\n\nfix: upgrading gevent\n\nfix: pep8 issue fixed for slack_notifier\n\nfix: change env list\n\nChange-Id: Iefc7fdc89a398623aa3515c89240b53a49c10282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/13121a1059492dc36a885e443a728b53cfafe3d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/13121a1059492dc36a885e443a728b53cfafe3d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
