)]}'
{"id":"openstack%2Fmonasca-notification~914469","triplet_id":"openstack%2Fmonasca-notification~master~Iac87df0e240964797ab34172210a27e419b4031c","project":"openstack/monasca-notification","branch":"master","topic":"reno-eom-victoria","hashtags":[],"change_id":"Iac87df0e240964797ab34172210a27e419b4031c","subject":"reno: Update master for unmaintained/victoria","status":"ABANDONED","created":"2024-03-27 10:45:54.000000000","updated":"2026-02-03 22:34:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0cbc1a89c17b2f1bdf777f978f8f5d2769798536","_number":914469,"virtual_id_number":914469,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-27 11:02:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-27 11:02:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-03 22:34:46.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"44350a8cd0906c1ef1790f85929e5988d34ff156","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-03-27 10:45:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a895c8ed1df6f1c49379dc432a2848d9c9aabf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 11:02:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dc17f2f03bf43889ab13a3014c1108f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/374a346f68d24a5887d3e496107a605e : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ddde2d2fb8846c4a03fa7af8b2c2267 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0cbc1a89c17b2f1bdf777f978f8f5d2769798536","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-03 22:34:46.000000000","message":"Abandoned\n\nThis project has been retired and no longer accepts code changes via https://review.opendev.org. New changes can be uploaded if the project is restored.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f2b840bf257d5eef7b496f03cba79682fc9cd9a","revisions":{"3f2b840bf257d5eef7b496f03cba79682fc9cd9a":{"kind":"REWORK","_number":1,"created":"2024-03-27 10:45:54.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/69/914469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/69/914469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/69/914469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/69/914469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/69/914469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/69/914469/1"}}},"commit":{"parents":[{"commit":"13121a1059492dc36a885e443a728b53cfafe3d8","subject":"Making the gates happy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/13121a1059492dc36a885e443a728b53cfafe3d8"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-27 10:45:53.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-27 10:45:53.000000000","tz":0},"subject":"reno: Update master for unmaintained/victoria","message":"reno: Update master for unmaintained/victoria\n\nUpdate the victoria release notes configuration to build from\nunmaintained/victoria.\n\nChange-Id: Iac87df0e240964797ab34172210a27e419b4031c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/3f2b840bf257d5eef7b496f03cba79682fc9cd9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/3f2b840bf257d5eef7b496f03cba79682fc9cd9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
