)]}'
{"id":"openstack%2Fmonasca-notification~919937","triplet_id":"openstack%2Fmonasca-notification~master~Ia462750553e3f0ab1a33d439508c4a08c3e71778","project":"openstack/monasca-notification","branch":"master","topic":"feat/upgrade","attention_set":{},"removed_from_attention_set":{"20865":{"account":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"last_update":"2024-05-19 18:33:38.000000000","reason":"\u003cGERRIT_ACCOUNT_20865\u003e replied on the change","reason_account":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}},"35988":{"account":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"last_update":"2024-06-04 08:17:03.000000000","reason":"Change was submitted"},"13114":{"account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"last_update":"2024-06-04 08:17:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia462750553e3f0ab1a33d439508c4a08c3e71778","subject":"fix orm to compatible with sqlalchemy 2","status":"MERGED","created":"2024-05-17 10:45:59.000000000","updated":"2024-06-04 08:17:03.000000000","submitted":"2024-06-04 08:17:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919937-feat/upgrade","meta_rev_id":"c10f0e000a1f38fdbe1cd50159d332fe54e70897","_number":919937,"virtual_id_number":919937,"owner":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-04 08:17:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"all":[{"value":2,"date":"2024-06-04 08:08:52.000000000","_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"all":[{"value":1,"date":"2024-06-04 08:08:52.000000000","_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"}],"reviewers":{"REVIEWER":[{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-17 11:15:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-18 11:04:06.000000000","updated_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"REVIEWER"},{"updated":"2024-05-19 14:32:49.000000000","updated_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"reviewer":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"state":"REVIEWER"}],"messages":[{"id":"f0cab7b32612541713a2d1e76371b5729e6e63fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-05-17 10:45:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5556164303eb5e52d619aef2d8c0869e568e499f","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-05-17 10:48:10.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"621704c7e1828d36801861da28783d3a9ef24479","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-05-17 10:48:44.000000000","message":"Removed Workflow+1 by \u003cGERRIT_ACCOUNT_13114\u003e\n","accounts_in_message":[{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}],"_revision_number":1},{"id":"53efde55e1c425634bc6fc1303f3fccb7ae39501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 11:15:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/650eea643b2d4aa6ada47ec0efcf67b2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/488fd2cb0c794ecf8aada47163148f15 : FAILURE in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01542332c99b4b28a1b360cca9a0d47b : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d400ded5cb2d420caf51f7abf74b4200 : FAILURE in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc7a203e3a0f4b2f92afc0f7f7bab2fc : FAILURE in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb485de49df54541bac6d272fadf3ed3 : FAILURE in 4m 06s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/932289fdec474926bfee745ba3d81828 : SUCCESS in 4m 29s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/ac2869fc7a29422ba3ba4a038be45027 : FAILURE in 27m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae761f0a1c05f861ce7728b82d95ef1e94f7dd85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-05-18 09:18:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2ac008b58ea1b74aac869ca02dfeb066cd6571d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 09:53:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a37e101181604431a84354ad10aa77dc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d23f4d15a7e42339435200830d73b49 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5326787b34004e758be88f0cb77c058f : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8319b732b8c4ed3bd9962601e6c69a8 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9e850e215d3478a8be127401384ede0 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02de0cabea1d439ea10f11f3ffecaac0 : SUCCESS in 3m 31s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/64a1db743b894e9f9d95e8de4d3caa58 : SUCCESS in 5m 53s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/d56b117cf7a049489f7ca9dd257e8e7b : FAILURE in 32m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1b5b364836ad76565ca8cebc8f1587e63066ff8f","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-05-18 11:04:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9661127aa4efcf445a5a79f1b342fbf90a1f0181","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-05-18 11:04:32.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6eda98001e0b2a1e73809d0c083b8c973bcf8a26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 11:29:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7477591108f4704a5f66ad4a3221e4e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/888c76ced3574cbb9989d7948ac2f96b : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f968aba329f34cbda281ebe1870de4cd : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a829a1eeb57541b982a09ec9e6dc09a3 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ce35038676d4391bac4e1dedd0ca8ef : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7215b332960d416fb347e00d2cc9e8a1 : SUCCESS in 2m 54s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/e021948f505344bcb66e8c9a0213f747 : SUCCESS in 4m 28s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/e13d6df9c88a408fb50150219443f305 : FAILURE in 23m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"78a6186e4cc08c401a609efb227a2152acdfbed7","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-05-19 18:33:38.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b65dd20fff8fc2aa44d65d0ceec7c573622a3289","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-06-03 23:57:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"80c474cd1819e53089602fac707da7ad0a440d75","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-06-04 05:57:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2cbe4bfdd6fdd4295cc963e13ebcacf29d61be5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 06:53:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/446d8ce10178446b8b1a8a815abe2918\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c8790e573964b7c8a75ba00f221ea27 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f01a3eea4b1e47d4ae543b82904e8715 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d556a4a664b4c28b611364ec4c9bde0 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d68f031b18447e5b646ee6e45feb847 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a288106f76404bbdb136765f3030258d : SUCCESS in 2m 42s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/b575c297c1374a129f9580bbfaf31f51 : SUCCESS in 4m 22s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/e0ac147f9a184ac6b640c9b4710ac041 : SUCCESS in 50m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"52f13c5f2e9e12082006282f46b72543f892b3f2","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-06-04 08:08:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e5a4960ffa1a529d158ca0e8cae5ba061501b12a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 08:09:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"94646f6818482603668a2843369bdeee81d741ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 08:17:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/259ace9fbdac4b03a75708a9a9066e70\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18abfe0d47b4466ebfc91b5692f0d20f : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2190110754fe41a7bffd291e37876b60 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29bca072c988454ca6fabc31669445ff : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":3},{"id":"c10f0e000a1f38fdbe1cd50159d332fe54e70897","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 08:17:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"841dcc0fb120e13198ca594c7f768d5b5fe6bba1","revisions":{"b9340e2299ec7a409734df4611df5a7455ec39dc":{"kind":"REWORK","_number":1,"created":"2024-05-17 10:45:59.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/37/919937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/37/919937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/37/919937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/37/919937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/37/919937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/37/919937/1"}}},"commit":{"parents":[{"commit":"13121a1059492dc36a885e443a728b53cfafe3d8","subject":"Making the gates happy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/13121a1059492dc36a885e443a728b53cfafe3d8"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-05-17 10:45:09.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-05-17 10:45:55.000000000","tz":420},"subject":"trigger zuul for test ci","message":"trigger zuul for test ci\n\nChange-Id: Ia462750553e3f0ab1a33d439508c4a08c3e71778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/b9340e2299ec7a409734df4611df5a7455ec39dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/b9340e2299ec7a409734df4611df5a7455ec39dc"}]},"branch":"refs/heads/master"},"25deb1aed5fac8117fa0cd1a168770ef2aad9a59":{"kind":"REWORK","_number":2,"created":"2024-05-18 09:18:10.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/37/919937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/37/919937/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/37/919937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/37/919937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/37/919937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/37/919937/2"}}},"commit":{"parents":[{"commit":"13121a1059492dc36a885e443a728b53cfafe3d8","subject":"Making the gates happy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/13121a1059492dc36a885e443a728b53cfafe3d8"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-05-17 10:45:09.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-05-18 09:14:42.000000000","tz":420},"subject":"fix orm to compatible with orm 2","message":"fix orm to compatible with orm 2\n\nChange-Id: Ia462750553e3f0ab1a33d439508c4a08c3e71778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/25deb1aed5fac8117fa0cd1a168770ef2aad9a59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/25deb1aed5fac8117fa0cd1a168770ef2aad9a59"}]},"branch":"refs/heads/master"},"841dcc0fb120e13198ca594c7f768d5b5fe6bba1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-05-18 11:04:32.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/37/919937/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/37/919937/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/37/919937/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/37/919937/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/37/919937/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/37/919937/3"}}},"commit":{"parents":[{"commit":"13121a1059492dc36a885e443a728b53cfafe3d8","subject":"Making the gates happy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/13121a1059492dc36a885e443a728b53cfafe3d8"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-05-17 10:45:09.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-05-18 11:04:32.000000000","tz":0},"subject":"fix orm to compatible with sqlalchemy 2","message":"fix orm to compatible with sqlalchemy 2\n\nChange-Id: Ia462750553e3f0ab1a33d439508c4a08c3e71778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/841dcc0fb120e13198ca594c7f768d5b5fe6bba1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/841dcc0fb120e13198ca594c7f768d5b5fe6bba1"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
