)]}'
{"id":"openstack%2Fmonasca-notification~957067","triplet_id":"openstack%2Fmonasca-notification~master~I14d11dba63a2cc980c21dbd6e6b7f204fc1f3556","project":"openstack/monasca-notification","branch":"master","topic":"retire-monasca-notification","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-11-24 13:51:10.000000000","reason":"Change was submitted"}},"hashtags":["monasca-retirement"],"change_id":"I14d11dba63a2cc980c21dbd6e6b7f204fc1f3556","subject":"Retire monasca-notification repository","status":"MERGED","created":"2025-08-12 05:18:59.000000000","updated":"2025-11-24 13:51:10.000000000","submitted":"2025-11-24 13:51:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957067-retire-monasca-notification","meta_rev_id":"6df91ae6c4abb44b5ca84bfb5c0358a2e6510ec2","_number":957067,"virtual_id_number":957067,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-24 13:51:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-11-24 13:49:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-08-19 17:54:16.000000000","_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"date":"2025-10-20 21:55:48.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2025-11-24 13:49:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-12 05:22:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-19 17:54:16.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-11-24 13:49:36.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"03728673cfe222eeea5e86636470630037b9f3f9","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-12 05:18:59.000000000","message":"Hashtag added: monasca-retirement","accounts_in_message":[],"_revision_number":1},{"id":"2450a62d1c5dbedea0a26994be8fd840d8dff4ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 05:22:05.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/953671 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"0604b3a1a2e8ef7087b3a6aba041afdafa159dbf","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-13 21:31:28.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4c0b2cd07650f76bcedd9afc0b52adedca7f4082","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-15 18:29:15.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"17b8b76f9d3eac3838f83ef1a5b346319290b67b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 18:29:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88d1718c20664ad788fec2e102626c6f\n\n- noop https://zuul.opendev.org/t/openstack/build/54f4ba2ba78f48bbbe3daf756e2eebe9 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"4571e73de4c9f513368a088e35605f051ae05465","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-08-19 17:54:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c89e6e583d81c9ec6e81b1f0db52caa5fbee316a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-20 17:01:26.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae8a7ca1fa59ba0c0ca5e798271a9a38a4061d17","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-20 21:00:22.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"2b2daa0f837330a15d2d490044eaa815b2cbdaf7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-20 21:55:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"712b0e2b77654070ed28583efabf8fee6c9e823a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 21:56:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4f07f4ab7bc46ce882412d90687f402\n\n- noop https://zuul.opendev.org/t/openstack/build/ccc40906a7ef43959d45986bd64f28cb : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"cb8afc6000ad9e7743dcee625eb10b3dbc67944e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-11-24 13:49:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"64a32262d1eb2d84de160747e546d83e6b3dc77f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 13:51:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eaf2bdba30fdb10a718e815003593840ce49c1b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 13:51:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09c6aeba6ca64689bc39891af21e0fa5\n\n- noop https://zuul.opendev.org/t/openstack/build/d35de1a488a34293be21710f0cd27532 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"6df91ae6c4abb44b5ca84bfb5c0358a2e6510ec2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 13:51:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a35f4d45da2d8b1082a2cd549f9e1ecff48baca6","revisions":{"f4b1a3b5f9575ba7f2b9e049e75adb8dcdbc1477":{"kind":"REWORK","_number":1,"created":"2025-08-12 05:18:59.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/67/957067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/67/957067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/67/957067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/67/957067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/67/957067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/67/957067/1"}}},"commit":{"parents":[{"commit":"841dcc0fb120e13198ca594c7f768d5b5fe6bba1","subject":"fix orm to compatible with sqlalchemy 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/841dcc0fb120e13198ca594c7f768d5b5fe6bba1"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-08-12 05:15:05.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-08-12 05:18:57.000000000","tz":-420},"subject":"Retire monasca-notification repository","message":"Retire monasca-notification repository\n\nThis repository is being retired as part of the Monasca project\nretirement. The project content has been replaced with a retirement\nnotice.\n\nDepends-On: I3cb522ce8f51424b64e93c1efaf0dfd1781cd5ac\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/957063\n\nChange-Id: I14d11dba63a2cc980c21dbd6e6b7f204fc1f3556\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f4b1a3b5f9575ba7f2b9e049e75adb8dcdbc1477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f4b1a3b5f9575ba7f2b9e049e75adb8dcdbc1477"}]},"branch":"refs/heads/master"},"a35f4d45da2d8b1082a2cd549f9e1ecff48baca6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-15 18:29:15.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/67/957067/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/67/957067/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/67/957067/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/67/957067/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/67/957067/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/67/957067/2"}}},"commit":{"parents":[{"commit":"841dcc0fb120e13198ca594c7f768d5b5fe6bba1","subject":"fix orm to compatible with sqlalchemy 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/841dcc0fb120e13198ca594c7f768d5b5fe6bba1"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-08-12 05:15:05.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-08-15 18:29:15.000000000","tz":0},"subject":"Retire monasca-notification repository","message":"Retire monasca-notification repository\n\nThis repository is being retired as part of the Monasca project\nretirement. The project content has been replaced with a retirement\nnotice.\n\nNeeded-By: I3cb522ce8f51424b64e93c1efaf0dfd1781cd5ac\nChange-Id: I14d11dba63a2cc980c21dbd6e6b7f204fc1f3556\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/a35f4d45da2d8b1082a2cd549f9e1ecff48baca6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/a35f4d45da2d8b1082a2cd549f9e1ecff48baca6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
