)]}'
{"id":"openstack%2Fmonasca-persister~770139","triplet_id":"openstack%2Fmonasca-persister~master~I43538b943176d81987db2f4ca139eb10f9b48385","project":"openstack/monasca-persister","branch":"master","topic":"test","hashtags":[],"change_id":"I43538b943176d81987db2f4ca139eb10f9b48385","subject":"Fix zuul publish docker image job","status":"MERGED","created":"2021-01-11 15:47:33.000000000","updated":"2021-01-18 14:57:57.000000000","submitted":"2021-01-18 14:57:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770139-test","meta_rev_id":"5ef2496f7278079004c97943945bf0abcd70225e","_number":770139,"virtual_id_number":770139,"owner":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-18 14:57:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-18 14:03:19.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-18 14:03:19.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-11 16:30:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-18 14:03:19.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"}],"messages":[{"id":"702ce5653c4ac6cf93e7092f409a1a25e803495b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2021-01-11 15:47:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b71777da35e651a15f14d9eeab9c92f16750d3f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 16:30:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbbe10d702334e7182ec12bad0930afb : FAILURE in 3m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6afb1305d2964d6abd6cff94f928e5b5 : FAILURE in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9d05e0d04bf486fbee58edc28c39360 : FAILURE in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21b88dee4415405889830dac5e6318af : FAILURE in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31202cf794bd4e278117c683084eec1a : FAILURE in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85ff52799e4043639b355f130e9ba7ef : FAILURE in 3m 27s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/667b2eddfd6f4ff984de182388f6e324 : SUCCESS in 38m 44s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/1163d8f8ea9b46339b40af850e7815b2 : FAILURE in 31m 10s (non-voting)\n- monasca-tempest-python3-java-cassandra https://zuul.opendev.org/t/openstack/build/525c12b1c36e4e328158b156923435d7 : FAILURE in 31m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"226be407a8c948a77cff7645674164fbf7ee215c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2021-01-11 16:57:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c3126acae0cb8c4d1de9bce2f4fb664397cb34a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 16:58:19.000000000","message":"Patch Set 2: Verified-1\n\nUnable to freeze job graph: Pre-review pipeline check does not allow post-review job publish-monasca-persister-docker-image","accounts_in_message":[],"_revision_number":2},{"id":"40300a0affeb57680fe8674f4d47b18fb451e3c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2021-01-12 15:38:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"80bc845063537c458d49181501596429642cac8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 15:38:40.000000000","message":"Patch Set 3: Verified-1\n\nUnable to freeze job graph: Pre-review pipeline check does not allow post-review job publish-monasca-persister-docker-image","accounts_in_message":[],"_revision_number":3},{"id":"ddca2058d00af156ce4e66af2cb59c74628e9e1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2021-01-14 12:39:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7694d41f96defb11c9bd61785f9c75120a04392e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:33:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2ad13f04e8f45d29e5a9de484bd9817 : FAILURE in 11m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c67ddff3e3fe44b9ba8080882990c699 : FAILURE in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/440ab01620284fb2bba66e73e482b6b2 : FAILURE in 11m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7ff54bd501d4a219657769df72245db : FAILURE in 9m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e291f6d5298e49c0a7f890665bd3c75e : FAILURE in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8d760da1ea9495292f94c5da950c4d9 : FAILURE in 11m 46s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/4dfbf79ec7d54bf6989de39747a1d8c3 : FAILURE in 35m 03s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/7580f3d3b30f4a078ca2d18def3439e5 : FAILURE in 33m 04s (non-voting)\n- monasca-tempest-python3-java-cassandra https://zuul.opendev.org/t/openstack/build/f79bb952103e498eb4cd5a7013bd792a : FAILURE in 33m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5173cecc8eb319add6f2fc2118d2a94ae8761bde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2021-01-18 13:11:31.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"8e69777419d31abf01c65638f1c15f37b186c64c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 14:00:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f47f93fe201469389a580b741740a25 : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/340b42fd968e4222b652ba258b268377 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a79cbe2022984098ad803277724dd46d : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff8b97e424124c79afb5f35a4fe8d8e0 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4ef0c1c41d142a2a4c380008e76262f : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2874e18906f04e08b818064393713322 : FAILURE in 12m 52s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/bb4635364df14296aaab7733b979e7c4 : SUCCESS in 41m 53s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/3bc85c91bdf74bf5b1bbb7ae3176857f : FAILURE in 32m 16s (non-voting)\n- monasca-tempest-python3-java-cassandra https://zuul.opendev.org/t/openstack/build/4431a015af994fe3b4080abc9908a629 : FAILURE in 34m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a4bfb7431583b112e947df22313de27908c52abd","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-01-18 14:03:19.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bea517bfc3f3d9a1e74e123f452c62fe11fc229a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 14:03:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f21b419c736891c08014acd2785e15d249fe5655","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 14:57:56.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/355e857e99504ed8ae367669f1d4c1ac : SUCCESS in 21m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/562f34ad1c3f493680e31d9cec8552db : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/255b100d813441e9bf83155a6adbcce1 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e84263924e94028819a8d588124df5d : SUCCESS in 5m 08s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/d07c932f6dc64bcfbef53db605e10ef5 : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":5},{"id":"5ef2496f7278079004c97943945bf0abcd70225e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 14:57:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"caac9719c76d7fcfd75960a1feb26e6e889f1c8a","revisions":{"52fee769d2b9efefcc4b4e28fd329bdf0424d5d7":{"kind":"REWORK","_number":1,"created":"2021-01-11 15:47:33.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/39/770139/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-persister","ref":"refs/changes/39/770139/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/1"}}},"commit":{"parents":[{"commit":"df0fd4958f19c34c3ade602a6f6e84d09c6122fe","subject":"Remove six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-persister/commit/df0fd4958f19c34c3ade602a6f6e84d09c6122fe"}]}],"author":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2021-01-11 15:47:06.000000000","tz":60},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2021-01-11 15:47:06.000000000","tz":60},"subject":"[WIP] Test change","message":"[WIP] Test change\n\nChange-Id: I43538b943176d81987db2f4ca139eb10f9b48385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-persister/commit/52fee769d2b9efefcc4b4e28fd329bdf0424d5d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-persister/commit/52fee769d2b9efefcc4b4e28fd329bdf0424d5d7"}]},"branch":"refs/heads/master"},"2d4f1f45cd4a607c322d74259a997e522a6eb0d4":{"kind":"REWORK","_number":2,"created":"2021-01-11 16:57:50.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/39/770139/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-persister","ref":"refs/changes/39/770139/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/2"}}},"commit":{"parents":[{"commit":"df0fd4958f19c34c3ade602a6f6e84d09c6122fe","subject":"Remove six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-persister/commit/df0fd4958f19c34c3ade602a6f6e84d09c6122fe"}]}],"author":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2021-01-11 15:47:06.000000000","tz":60},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2021-01-11 16:57:38.000000000","tz":60},"subject":"[WIP] Test change","message":"[WIP] Test change\n\nChange-Id: I43538b943176d81987db2f4ca139eb10f9b48385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-persister/commit/2d4f1f45cd4a607c322d74259a997e522a6eb0d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-persister/commit/2d4f1f45cd4a607c322d74259a997e522a6eb0d4"}]},"branch":"refs/heads/master"},"3b63527c91d15bd596d37022f7c07c2b71b51477":{"kind":"REWORK","_number":3,"created":"2021-01-12 15:38:15.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/39/770139/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-persister","ref":"refs/changes/39/770139/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/3"}}},"commit":{"parents":[{"commit":"df0fd4958f19c34c3ade602a6f6e84d09c6122fe","subject":"Remove six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-persister/commit/df0fd4958f19c34c3ade602a6f6e84d09c6122fe"}]}],"author":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2021-01-11 15:47:06.000000000","tz":60},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2021-01-12 15:37:41.000000000","tz":60},"subject":"[WIP] Test change","message":"[WIP] Test change\n\nChange-Id: I43538b943176d81987db2f4ca139eb10f9b48385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-persister/commit/3b63527c91d15bd596d37022f7c07c2b71b51477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-persister/commit/3b63527c91d15bd596d37022f7c07c2b71b51477"}]},"branch":"refs/heads/master"},"850478dee97cd93fe47cfdc96b7b02e61fc804a0":{"kind":"REWORK","_number":4,"created":"2021-01-14 12:39:52.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/39/770139/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-persister","ref":"refs/changes/39/770139/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/4"}}},"commit":{"parents":[{"commit":"df0fd4958f19c34c3ade602a6f6e84d09c6122fe","subject":"Remove six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-persister/commit/df0fd4958f19c34c3ade602a6f6e84d09c6122fe"}]}],"author":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2021-01-11 15:47:06.000000000","tz":60},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2021-01-14 12:37:40.000000000","tz":60},"subject":"Fix zuul publish docker image job","message":"Fix zuul publish docker image job\n\nAdd tag to docker push commend\n\nChange-Id: I43538b943176d81987db2f4ca139eb10f9b48385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-persister/commit/850478dee97cd93fe47cfdc96b7b02e61fc804a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-persister/commit/850478dee97cd93fe47cfdc96b7b02e61fc804a0"}]},"branch":"refs/heads/master"},"caac9719c76d7fcfd75960a1feb26e6e889f1c8a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-01-18 13:11:31.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/39/770139/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-persister","ref":"refs/changes/39/770139/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-persister refs/changes/39/770139/5"}}},"commit":{"parents":[{"commit":"37411e792a8ee36eb04832785044b048719e9fcf","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-persister/commit/37411e792a8ee36eb04832785044b048719e9fcf"}]}],"author":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2021-01-11 15:47:06.000000000","tz":60},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2021-01-18 13:11:31.000000000","tz":0},"subject":"Fix zuul publish docker image job","message":"Fix zuul publish docker image job\n\nAdd tag to docker push commend\n\nChange-Id: I43538b943176d81987db2f4ca139eb10f9b48385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-persister/commit/caac9719c76d7fcfd75960a1feb26e6e889f1c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-persister/commit/caac9719c76d7fcfd75960a1feb26e6e889f1c8a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
