)]}'
{"id":"openstack%2Fmonasca-specs~547233","triplet_id":"openstack%2Fmonasca-specs~master~I915376827604bc692cd26b7ed00812c64ee2e3c0","project":"openstack/monasca-specs","branch":"master","topic":"547233","hashtags":[],"change_id":"I915376827604bc692cd26b7ed00812c64ee2e3c0","subject":"Metrics retention policy enhancement","status":"MERGED","created":"2018-02-23 01:51:31.000000000","updated":"2019-10-14 17:35:14.000000000","submitted":"2019-10-14 17:27:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"547233-1571074052106-70f2cce9","meta_rev_id":"e759a47c86beb201caf4e2cd6fb5b2099d14dba4","_number":547233,"virtual_id_number":547233,"owner":{"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":9320,"name":"Dan Dyer","email":"dan.dyer00@gmail.com","username":"dandyer00"},{"value":0,"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-14 17:27:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11580,"name":"Ashwin Agate","email":"agateaaa@gmail.com","username":"agateaaa"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"all":[{"value":0,"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},{"value":2,"date":"2018-11-01 16:29:44.000000000","_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":9320,"name":"Dan Dyer","email":"dan.dyer00@gmail.com","username":"dandyer00"},{"value":0,"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11580,"name":"Ashwin Agate","email":"agateaaa@gmail.com","username":"agateaaa"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"all":[{"value":0,"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},{"value":1,"date":"2019-10-14 17:10:44.000000000","_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":9320,"name":"Dan Dyer","email":"dan.dyer00@gmail.com","username":"dandyer00"},{"value":0,"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11580,"name":"Ashwin Agate","email":"agateaaa@gmail.com","username":"agateaaa"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9320,"name":"Dan Dyer","email":"dan.dyer00@gmail.com","username":"dandyer00"},{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":11580,"name":"Ashwin Agate","email":"agateaaa@gmail.com","username":"agateaaa"},{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-02-23 19:55:06.000000000","updated_by":{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},"reviewer":{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},"state":"REVIEWER"},{"updated":"2018-02-23 19:55:56.000000000","updated_by":{"_account_id":11580,"name":"Ashwin Agate","email":"agateaaa@gmail.com","username":"agateaaa"},"reviewer":{"_account_id":11580,"name":"Ashwin Agate","email":"agateaaa@gmail.com","username":"agateaaa"},"state":"REVIEWER"},{"updated":"2018-02-28 18:31:29.000000000","updated_by":{"_account_id":9320,"name":"Dan Dyer","email":"dan.dyer00@gmail.com","username":"dandyer00"},"reviewer":{"_account_id":9320,"name":"Dan Dyer","email":"dan.dyer00@gmail.com","username":"dandyer00"},"state":"REVIEWER"},{"updated":"2018-03-20 22:00:48.000000000","updated_by":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"reviewer":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"state":"REVIEWER"},{"updated":"2018-09-27 09:21:18.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-10-14 17:10:44.000000000","updated_by":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"reviewer":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"state":"REVIEWER"},{"updated":"2019-10-14 17:27:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42897311244f47c1a84c90bfa394f155beee8728","author":{"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},"date":"2018-02-23 01:51:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e1d4cc9b19083a6d94e549389f73c9c5451803f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-23 01:59:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/33/547233/1/check/build-openstack-sphinx-docs/960a291/ : FAILURE in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"00384aeb999c8e69d09b896929ff0c2b5fc46d8a","author":{"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},"date":"2018-02-23 02:42:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c7029a06e6248399fbee3daeea90b4f759253ab8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-23 02:47:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/33/547233/2/check/build-openstack-sphinx-docs/ac56c08/ : FAILURE in 3m 34s","accounts_in_message":[],"_revision_number":2},{"id":"5a99260f0d14523303634712c82a59e3d7a77189","author":{"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},"date":"2018-02-23 03:36:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"078c5815d83e802090d77dfcb5e4e553e5ec7be2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-23 03:40:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/33/547233/3/check/build-openstack-sphinx-docs/11847bd/ : FAILURE in 3m 29s","accounts_in_message":[],"_revision_number":3},{"id":"2be9814ae58b7419c6b8a72879d195fef4944906","author":{"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},"date":"2018-02-23 17:24:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"58c711e25cab4f0edf4461ffcb2fa9d1f3edbe5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-23 17:30:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/33/547233/4/check/build-openstack-sphinx-docs/5a3e911/ : FAILURE in 4m 34s","accounts_in_message":[],"_revision_number":4},{"id":"cba1a8c981952a091755d7c8e44bdfb65babf3fb","author":{"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},"date":"2018-02-23 19:32:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b3778916fe3680bee11d5c10020a6f08b5bf6540","author":{"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},"date":"2018-02-28 08:39:20.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cbef3e41c0cae524ec4fa149d6aaeb5eb9553e49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-28 08:44:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/33/547233/6/check/build-openstack-sphinx-docs/8e82425/html/ : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":6},{"id":"2a2f9727f77524c6d276e4d96795e4e2110cc6f0","author":{"_account_id":9320,"name":"Dan Dyer","email":"dan.dyer00@gmail.com","username":"dandyer00"},"date":"2018-02-28 18:31:29.000000000","message":"Patch Set 6:\n\n(7 comments)\n\nOverall I think we need to look at the metrics retention more from a metrics behavior and functional perspective. Would also like to see a more consolidated policy approach that is configurable via API. and maybe a pink pony as well :)","accounts_in_message":[],"_revision_number":6},{"id":"285a8fc4aeb7b02389e1a2a9be1ba3ad88679b55","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-03-20 14:46:20.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"b9c16bbd8844e92ec5ab7006f9b9c7c58553ad15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-20 14:57:14.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/33/547233/7/check/build-openstack-sphinx-docs/dd7055f/ : FAILURE in 3m 51s","accounts_in_message":[],"_revision_number":7},{"id":"667adaedb4059eabfb5397dca7495a42ab9740b3","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-03-20 15:00:51.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6d029c092db3883cc50a6733f9bf7c19535ea78d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-20 15:09:41.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/33/547233/8/check/build-openstack-sphinx-docs/f7e523e/ : FAILURE in 3m 42s","accounts_in_message":[],"_revision_number":8},{"id":"12041e2634ffb74304985533ca3510b4574ec85b","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-03-20 22:00:48.000000000","message":"Patch Set 8:\n\n(8 comments)\n\nCheck trailing white spaces.","accounts_in_message":[],"_revision_number":8},{"id":"a46a1885d6b972f9af185991f8649ad8499f75f0","author":{"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},"date":"2018-03-22 05:49:50.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"fb849cf17da2376938843b5f5b5cdafae5b8e91a","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-09-07 21:37:40.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"fe6e7f4ef2bc896f3af11fba7d4a8c66dd48c67c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-07 21:47:20.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/33/547233/9/check/openstack-tox-docs/3fcb0b2/ : FAILURE in 3m 52s","accounts_in_message":[],"_revision_number":9},{"id":"dd3bfd55405d7e6d63c6fd94e70dfdbaf5d8238d","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-09-07 21:59:19.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7ccd1d6f06af3c1c6bab15ca1268ffbb00aefe72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-07 22:03:20.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/33/547233/10/check/openstack-tox-docs/e541eda/ : FAILURE in 3m 40s","accounts_in_message":[],"_revision_number":10},{"id":"2c611513e863b76d0b1c2d7a74ff391f884dfad8","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-09-07 22:07:56.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"5f8a8fdfa2578e44f86e1176379f52c4e58f19b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-07 22:21:53.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/33/547233/11/check/openstack-tox-docs/6b19819/ : FAILURE in 4m 14s","accounts_in_message":[],"_revision_number":11},{"id":"7711f7acaa2cf0e259cefe17c5b9b3ff23792106","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-09-07 23:21:38.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"642b026dcbc262b62f0207643d33b62e4b045c5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-07 23:26:48.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/33/547233/12/check/openstack-tox-docs/7a3a5d6/ : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":12},{"id":"db5ef7b906b373546535695f3eafea6eca7b3859","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-09-07 23:28:56.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"7bf3364b9569404526a70472e35854901308707b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-07 23:37:28.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/547233/13/check/openstack-tox-docs/1d76063/html/ : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":13},{"id":"2cef404b868d3d57f9f28a2694ffcd4b014e24b4","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-09-08 04:11:10.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"05eb48c10a2af4a15ef47693ec1afa828071a84c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-08 04:16:30.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/547233/14/check/openstack-tox-docs/017c671/html/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":14},{"id":"593890ffa2f4b6acee4d02e5e0fcd22f9f2979f7","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-09-11 21:34:33.000000000","message":"Patch Set 14: Workflow-1\n\nMajor revision in the works. Watch this space","accounts_in_message":[],"_revision_number":14},{"id":"5e80bdc4f95de7cf9f87cd48095b5a3179a63f45","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-09-13 22:01:55.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"dd86d7d35d4780a23cf61011b5e99226953452c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-13 22:08:46.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/33/547233/15/check/openstack-tox-docs/7edfb19/ : FAILURE in 5m 28s","accounts_in_message":[],"_revision_number":15},{"id":"455d548e315d6ceedf22198b543534cdabf90875","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-09-26 21:23:38.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"eab983bd1435c105e5e6d1a251d70f46ba916ed2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-27 03:31:49.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/33/547233/16/check/openstack-tox-docs/fc61d39/ : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":16},{"id":"5f5c2ba8507fa0512165679b15083335150af40e","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-09-27 09:21:18.000000000","message":"Patch Set 16: Code-Review-1\n\nPlease split this to two separate changes.","accounts_in_message":[],"_revision_number":16},{"id":"df78d04f771d5652ac9b4cc42b6d33d5e5bc8736","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-09-27 15:25:59.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"a6daab8837dac7ef22f0c6343fb6a92aed57d551","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-09-27 15:31:20.000000000","message":"Patch Set 17: Code-Review+1\n\nDone - https://review.openstack.org/#/c/605782","accounts_in_message":[],"_revision_number":17},{"id":"dab09f9b00203d3e8974060c459242dc856e415f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-27 20:41:31.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/33/547233/17/check/openstack-tox-docs/0b54f7f/ : FAILURE in 4m 14s","accounts_in_message":[],"_revision_number":17},{"id":"6bdef91f277da4ee53873feac0c3cc29447e6f10","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-09-27 22:47:50.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"58438140b68c4f4692d3ea752ae3b6808d0f5b5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-28 03:01:10.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/33/547233/18/check/openstack-tox-docs/b1cf326/ : FAILURE in 4m 43s","accounts_in_message":[],"_revision_number":18},{"id":"820ba1db1a9ca1607ab7e1763edb34ac289ac53d","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-09-28 20:09:09.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"06260ea27283b206915463a7f5c01ba5ebdee298","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-28 22:01:52.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/547233/19/check/openstack-tox-docs/dc16d1c/html/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":19},{"id":"959a966095708b53d548069f15fb43f0bfd192fb","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-10-26 22:29:31.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"21726283e57b30cbc07e0dd3f930e7760e851f6c","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-10-26 22:37:20.000000000","message":"Patch Set 20: Code-Review+1\n\nI think this is in good shape to merge.","accounts_in_message":[],"_revision_number":20},{"id":"895d1121d8e061caef896b3ed9b4334fd1e3d421","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-27 02:21:12.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/547233/20/check/openstack-tox-docs/b1f52b9/html/ : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":20},{"id":"c0e85a73c1aed1eaeba9db9c92ecd3c9f5f23d3b","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-11-01 16:29:44.000000000","message":"Patch Set 20: Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"0d69580c8a91cb0d30b0d03fa8348afb9fefec4f","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-10-14 17:10:44.000000000","message":"Patch Set 20: Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"be3534cba3831cbe5c953a3787167e49d9dc7344","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:11:05.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"eab56d96fb9422ae001a4d23f28f2368d1fe59f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:27:31.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/758360db1dc84135ae4a68470dcaa3b6 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":20},{"id":"16adad92c743dcedfed25c847ce174c3e5f859c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:27:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":20},{"id":"7be9d855df07af07f20c1075a47da28844fb7570","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:35:14.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/66f6a52680ee435a97c1748649defb54 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"a3f82ddd4aa073897b8e83ad162ccba2012472d5","revisions":{"0b9f927f8c52aa0e6cc653c61b9350fdb431db1f":{"kind":"REWORK","_number":1,"created":"2018-02-23 01:51:31.000000000","uploader":{"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},"ref":"refs/changes/33/547233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/1"}}},"commit":{"parents":[{"commit":"05be23e6d9e244b17ddf481cae4399767712bdea","subject":"Add chandankumar to split_tempest goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/05be23e6d9e244b17ddf481cae4399767712bdea"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:51:30.000000000","tz":-480},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupprot differentiable metrics retention policy based on metrics\ntype.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/0b9f927f8c52aa0e6cc653c61b9350fdb431db1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/0b9f927f8c52aa0e6cc653c61b9350fdb431db1f"}]},"branch":"refs/heads/master"},"8375c83c2083286adfaf1607b91d226b94d8c5fa":{"kind":"REWORK","_number":2,"created":"2018-02-23 02:42:22.000000000","uploader":{"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},"ref":"refs/changes/33/547233/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/2"}}},"commit":{"parents":[{"commit":"05be23e6d9e244b17ddf481cae4399767712bdea","subject":"Add chandankumar to split_tempest goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/05be23e6d9e244b17ddf481cae4399767712bdea"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 02:42:16.000000000","tz":-480},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupprot differentiable metrics retention policy based on metrics\ntype.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/8375c83c2083286adfaf1607b91d226b94d8c5fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/8375c83c2083286adfaf1607b91d226b94d8c5fa"}]},"branch":"refs/heads/master"},"da175d6b044382b956ab629933322515c7d0a573":{"kind":"REWORK","_number":3,"created":"2018-02-23 03:36:24.000000000","uploader":{"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},"ref":"refs/changes/33/547233/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/3"}}},"commit":{"parents":[{"commit":"05be23e6d9e244b17ddf481cae4399767712bdea","subject":"Add chandankumar to split_tempest goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/05be23e6d9e244b17ddf481cae4399767712bdea"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 03:36:16.000000000","tz":-480},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupprot differentiable metrics retention policy based on metrics\ntype.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/da175d6b044382b956ab629933322515c7d0a573"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/da175d6b044382b956ab629933322515c7d0a573"}]},"branch":"refs/heads/master"},"b0027c7846d2c7f88414c3c14f584a39fe4f77f7":{"kind":"REWORK","_number":4,"created":"2018-02-23 17:24:22.000000000","uploader":{"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},"ref":"refs/changes/33/547233/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/4"}}},"commit":{"parents":[{"commit":"05be23e6d9e244b17ddf481cae4399767712bdea","subject":"Add chandankumar to split_tempest goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/05be23e6d9e244b17ddf481cae4399767712bdea"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 17:24:14.000000000","tz":-480},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupprot differentiable metrics retention policy based on metrics\ntype.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/b0027c7846d2c7f88414c3c14f584a39fe4f77f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/b0027c7846d2c7f88414c3c14f584a39fe4f77f7"}]},"branch":"refs/heads/master"},"3778ca6d42003555c90c372f3274cade5b76a4c1":{"kind":"REWORK","_number":5,"created":"2018-02-23 19:32:11.000000000","uploader":{"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},"ref":"refs/changes/33/547233/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/5"}}},"commit":{"parents":[{"commit":"05be23e6d9e244b17ddf481cae4399767712bdea","subject":"Add chandankumar to split_tempest goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/05be23e6d9e244b17ddf481cae4399767712bdea"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 19:32:03.000000000","tz":-480},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupprot differentiable metrics retention policy based on metrics\ntype.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/3778ca6d42003555c90c372f3274cade5b76a4c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/3778ca6d42003555c90c372f3274cade5b76a4c1"}]},"branch":"refs/heads/master"},"e2115e21efaf98ed06df1ffe4ffa487e11c29b13":{"kind":"REWORK","_number":6,"created":"2018-02-28 08:39:20.000000000","uploader":{"_account_id":24580,"name":"James Gu","email":"james.gu@microsoft.com","username":"jamesgu"},"ref":"refs/changes/33/547233/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/6"}}},"commit":{"parents":[{"commit":"05be23e6d9e244b17ddf481cae4399767712bdea","subject":"Add chandankumar to split_tempest goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/05be23e6d9e244b17ddf481cae4399767712bdea"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-28 08:39:00.000000000","tz":-480},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupprot differentiable metrics retention policy based on metrics\ntype.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/e2115e21efaf98ed06df1ffe4ffa487e11c29b13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/e2115e21efaf98ed06df1ffe4ffa487e11c29b13"}]},"branch":"refs/heads/master"},"423b8aa17122286fbb735d1ea4c0cca3c11df4b2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2018-03-20 14:46:20.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/33/547233/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/7"}}},"commit":{"parents":[{"commit":"00594545a42ad0d4e477a74017e619f244a242df","subject":"Update directory structure for Rocky specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/00594545a42ad0d4e477a74017e619f244a242df"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"Witold Bedyk","email":"witold.bedyk@est.fujitsu.com","date":"2018-03-20 14:46:20.000000000","tz":0},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupprot differentiable metrics retention policy based on metrics\ntype.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/423b8aa17122286fbb735d1ea4c0cca3c11df4b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/423b8aa17122286fbb735d1ea4c0cca3c11df4b2"}]},"branch":"refs/heads/master"},"33c489665a09db058cc5c86b45962feabdc19d3c":{"kind":"REWORK","_number":8,"created":"2018-03-20 15:00:51.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/33/547233/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/8"}}},"commit":{"parents":[{"commit":"00594545a42ad0d4e477a74017e619f244a242df","subject":"Update directory structure for Rocky specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/00594545a42ad0d4e477a74017e619f244a242df"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"Witold Bedyk","email":"witold.bedyk@est.fujitsu.com","date":"2018-03-20 14:59:34.000000000","tz":60},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupport differentiable metrics retention policy based on metrics\ntype.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/33c489665a09db058cc5c86b45962feabdc19d3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/33c489665a09db058cc5c86b45962feabdc19d3c"}]},"branch":"refs/heads/master"},"ca82e7f34a9c32ad7a68590c8e966b43f80d2eef":{"kind":"REWORK","_number":9,"created":"2018-09-07 21:37:40.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/33/547233/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/9"}}},"commit":{"parents":[{"commit":"4aa92c0caa564a0f7bde40e6630df8edc76e042d","subject":"import zuul job settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/4aa92c0caa564a0f7bde40e6630df8edc76e042d"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-09-07 21:37:30.000000000","tz":-420},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupport differentiable metrics retention policy based on metrics\ntype.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/ca82e7f34a9c32ad7a68590c8e966b43f80d2eef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/ca82e7f34a9c32ad7a68590c8e966b43f80d2eef"}]},"branch":"refs/heads/master"},"d04fadf25d898bd97789b3efb1c3d6aa5051372b":{"kind":"REWORK","_number":10,"created":"2018-09-07 21:59:19.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/33/547233/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/10"}}},"commit":{"parents":[{"commit":"4aa92c0caa564a0f7bde40e6630df8edc76e042d","subject":"import zuul job settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/4aa92c0caa564a0f7bde40e6630df8edc76e042d"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-09-07 21:59:10.000000000","tz":-420},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupport differentiable metrics retention policy based on metrics\ntype.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/d04fadf25d898bd97789b3efb1c3d6aa5051372b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/d04fadf25d898bd97789b3efb1c3d6aa5051372b"}]},"branch":"refs/heads/master"},"3fcec81b5a86e861dd9a123bafe1efa00d437ce3":{"kind":"REWORK","_number":11,"created":"2018-09-07 22:07:56.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/33/547233/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/11"}}},"commit":{"parents":[{"commit":"4aa92c0caa564a0f7bde40e6630df8edc76e042d","subject":"import zuul job settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/4aa92c0caa564a0f7bde40e6630df8edc76e042d"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-09-07 22:07:35.000000000","tz":-420},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupport differentiable metrics retention policy based on metrics\ntype.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/3fcec81b5a86e861dd9a123bafe1efa00d437ce3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/3fcec81b5a86e861dd9a123bafe1efa00d437ce3"}]},"branch":"refs/heads/master"},"1cac22d9e39515f8915e3c04ff14fd0d92cd7bef":{"kind":"REWORK","_number":12,"created":"2018-09-07 23:21:38.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/33/547233/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/12"}}},"commit":{"parents":[{"commit":"4aa92c0caa564a0f7bde40e6630df8edc76e042d","subject":"import zuul job settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/4aa92c0caa564a0f7bde40e6630df8edc76e042d"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-09-07 23:21:26.000000000","tz":-420},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupport differentiable metrics retention policy based on metrics\ntype.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/1cac22d9e39515f8915e3c04ff14fd0d92cd7bef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/1cac22d9e39515f8915e3c04ff14fd0d92cd7bef"}]},"branch":"refs/heads/master"},"d3d4f84a0b45d24f710a43c8d55451b8265107f2":{"kind":"REWORK","_number":13,"created":"2018-09-07 23:28:56.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/33/547233/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/13"}}},"commit":{"parents":[{"commit":"4aa92c0caa564a0f7bde40e6630df8edc76e042d","subject":"import zuul job settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/4aa92c0caa564a0f7bde40e6630df8edc76e042d"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-09-07 23:28:47.000000000","tz":-420},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupport differentiable metrics retention policy based on metrics\ntype.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/d3d4f84a0b45d24f710a43c8d55451b8265107f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/d3d4f84a0b45d24f710a43c8d55451b8265107f2"}]},"branch":"refs/heads/master"},"f3501176e223ca179e2e388dcd3d809aee785764":{"kind":"REWORK","_number":14,"created":"2018-09-08 04:11:10.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/33/547233/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/14"}}},"commit":{"parents":[{"commit":"4aa92c0caa564a0f7bde40e6630df8edc76e042d","subject":"import zuul job settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/4aa92c0caa564a0f7bde40e6630df8edc76e042d"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-09-08 04:10:41.000000000","tz":-420},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupport differentiable metrics retention policy based on metrics\ntype.  Also outline an alternative (that may be preferable).\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/f3501176e223ca179e2e388dcd3d809aee785764"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/f3501176e223ca179e2e388dcd3d809aee785764"}]},"branch":"refs/heads/master"},"744f5dc63914f37bb0c4f0dc4b2256292862b249":{"kind":"REWORK","_number":15,"created":"2018-09-13 22:01:55.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/33/547233/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/15"}}},"commit":{"parents":[{"commit":"4aa92c0caa564a0f7bde40e6630df8edc76e042d","subject":"import zuul job settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/4aa92c0caa564a0f7bde40e6630df8edc76e042d"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-09-13 22:01:13.000000000","tz":-420},"subject":"WIP Metrics retention policy enhancement","message":"WIP Metrics retention policy enhancement\n\nSupport differentiable metrics retention policy based on metrics\ntype.  Also outline alternatives.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/744f5dc63914f37bb0c4f0dc4b2256292862b249"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/744f5dc63914f37bb0c4f0dc4b2256292862b249"}]},"branch":"refs/heads/master"},"b2ae09065c4323dc58e26433fd1adf5b74823e68":{"kind":"REWORK","_number":16,"created":"2018-09-26 21:23:38.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/33/547233/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/16"}}},"commit":{"parents":[{"commit":"4aa92c0caa564a0f7bde40e6630df8edc76e042d","subject":"import zuul job settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/4aa92c0caa564a0f7bde40e6630df8edc76e042d"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-09-26 21:21:01.000000000","tz":-420},"subject":"Metrics retention policy enhancement, python persister perf","message":"Metrics retention policy enhancement, python persister perf\n\nSupport differentiable metrics retention policy based on metrics\ntype.  Also outline alternatives.\n\nThis commit also includes a spec for Python Persiseter Performance\nmetric collection, still Work in Progress.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/b2ae09065c4323dc58e26433fd1adf5b74823e68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/b2ae09065c4323dc58e26433fd1adf5b74823e68"}]},"branch":"refs/heads/master"},"c1ca4a1015082fe1dd3d58ffc579b7c931c6aaeb":{"kind":"REWORK","_number":17,"created":"2018-09-27 15:25:59.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/33/547233/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/17"}}},"commit":{"parents":[{"commit":"4aa92c0caa564a0f7bde40e6630df8edc76e042d","subject":"import zuul job settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/4aa92c0caa564a0f7bde40e6630df8edc76e042d"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-09-27 15:25:15.000000000","tz":-420},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupport differentiable metrics retention policy based on metrics\ntype.  Also outline alternatives.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/c1ca4a1015082fe1dd3d58ffc579b7c931c6aaeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/c1ca4a1015082fe1dd3d58ffc579b7c931c6aaeb"}]},"branch":"refs/heads/master"},"867f2dfe17d2c23a2e0eba796a5907f80572680a":{"kind":"REWORK","_number":18,"created":"2018-09-27 22:47:50.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/33/547233/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/18"}}},"commit":{"parents":[{"commit":"4aa92c0caa564a0f7bde40e6630df8edc76e042d","subject":"import zuul job settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/4aa92c0caa564a0f7bde40e6630df8edc76e042d"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-09-27 22:47:38.000000000","tz":-420},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupport differentiable metrics retention policy based on metrics\ntype.  Also outline alternatives.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/867f2dfe17d2c23a2e0eba796a5907f80572680a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/867f2dfe17d2c23a2e0eba796a5907f80572680a"}]},"branch":"refs/heads/master"},"13ae5571fcff6126e656fca3344ae87e00d15125":{"kind":"REWORK","_number":19,"created":"2018-09-28 20:09:09.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/33/547233/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/19"}}},"commit":{"parents":[{"commit":"4aa92c0caa564a0f7bde40e6630df8edc76e042d","subject":"import zuul job settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/4aa92c0caa564a0f7bde40e6630df8edc76e042d"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-09-28 20:08:52.000000000","tz":-420},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupport differentiable metrics retention policy based on metrics\ntype.  Also outline alternatives.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/13ae5571fcff6126e656fca3344ae87e00d15125"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/13ae5571fcff6126e656fca3344ae87e00d15125"}]},"branch":"refs/heads/master"},"a3f82ddd4aa073897b8e83ad162ccba2012472d5":{"kind":"REWORK","_number":20,"created":"2018-10-26 22:29:31.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/33/547233/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/33/547233/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/33/547233/20"}}},"commit":{"parents":[{"commit":"4aa92c0caa564a0f7bde40e6630df8edc76e042d","subject":"import zuul job settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/4aa92c0caa564a0f7bde40e6630df8edc76e042d"}]}],"author":{"name":"James Gu","email":"jgu@suse.com","date":"2018-02-23 01:49:49.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-10-26 22:29:13.000000000","tz":-420},"subject":"Metrics retention policy enhancement","message":"Metrics retention policy enhancement\n\nSupport differentiable metrics retention policy based on metrics\ntype.  Also outline alternatives.\n\nChange-Id: I915376827604bc692cd26b7ed00812c64ee2e3c0\nstory: 2001576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/a3f82ddd4aa073897b8e83ad162ccba2012472d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/a3f82ddd4aa073897b8e83ad162ccba2012472d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
