)]}'
{"id":"openstack%2Fmonasca-specs~660459","triplet_id":"openstack%2Fmonasca-specs~master~I152a6eaf5f20bb1a271a8ccf3fba5cbd30893f68","project":"openstack/monasca-specs","branch":"master","topic":"660459","hashtags":[],"change_id":"I152a6eaf5f20bb1a271a8ccf3fba5cbd30893f68","subject":"Add Train priorities page","status":"MERGED","created":"2019-05-21 17:11:19.000000000","updated":"2019-06-07 09:13:15.000000000","submitted":"2019-06-07 09:13:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660459-1559898795853-d9d96933","meta_rev_id":"9f53b31bdf37da84bc43eb06ba9df0de1566f62b","_number":660459,"virtual_id_number":660459,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":2,"date":"2019-06-07 09:13:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-06-07 09:04:43.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-05-22 14:16:59.000000000","_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-06-07 09:04:43.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 14:16:59.000000000","updated_by":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"reviewer":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"state":"REVIEWER"},{"updated":"2019-06-07 09:04:43.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-06-07 09:13:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e113847d03cfb0d8dbd05a4cd213427a6b89dbd","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-05-21 17:11:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96d795929affb68a9aa38d8e716e8b026d4c1220","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 17:20:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/59/660459/1/check/openstack-tox-docs/4bd25e7/ : FAILURE in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ea30d2c99943aa32506ae8954a7c6310221849ea","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-05-21 17:39:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nReally minor typo. :)","accounts_in_message":[],"_revision_number":1},{"id":"ce07972630a124e0f1cb430bb9e5b8470f822725","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-05-22 08:43:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"74eb8fe4bae2630847c7d14163a824bd0f2d4d29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 08:51:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/59/660459/2/check/openstack-tox-docs/c140ad3/html/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"957944e3980391d3645e6970d51720d64ded674e","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-05-22 14:16:59.000000000","message":"Patch Set 2: Code-Review+2\n\nall good","accounts_in_message":[],"_revision_number":2},{"id":"59e51f1f44043b5696588e50de0bf12ea06d3162","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-06-07 09:04:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks all","accounts_in_message":[],"_revision_number":2},{"id":"1299e658a68462614d2319aa17da95834910b4ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 09:04:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9844eac3b9ac477b4561ba5ce5ddd20862a60568","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 09:13:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/59/660459/2/gate/openstack-tox-docs/63f7cca/html/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"50da67e8321ab6b5264cec8e6525b355e7ae463a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 09:13:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ae237f9307192759dcb685ed4ab741c12ab99bb9","revisions":{"8bf6f10d2e81318f959f9d4f4ce811895acd27b2":{"kind":"REWORK","_number":1,"created":"2019-05-21 17:11:19.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/59/660459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/59/660459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/59/660459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/59/660459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/59/660459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/59/660459/1"}}},"commit":{"parents":[{"commit":"ab91662dab330e67ff6b1709319b98d52006f853","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/ab91662dab330e67ff6b1709319b98d52006f853"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-05-21 17:10:53.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-05-21 17:11:17.000000000","tz":120},"subject":"Add Train priorities page","message":"Add Train priorities page\n\nChange-Id: I152a6eaf5f20bb1a271a8ccf3fba5cbd30893f68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/8bf6f10d2e81318f959f9d4f4ce811895acd27b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/8bf6f10d2e81318f959f9d4f4ce811895acd27b2"}]},"branch":"refs/heads/master"},"ae237f9307192759dcb685ed4ab741c12ab99bb9":{"kind":"REWORK","_number":2,"created":"2019-05-22 08:43:20.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/59/660459/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/59/660459/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/59/660459/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/59/660459/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/59/660459/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/59/660459/2"}}},"commit":{"parents":[{"commit":"8e772475d34007bfdeccc38dac36ed5fd3148139","subject":"Resolve issues with Sphinx 2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/8e772475d34007bfdeccc38dac36ed5fd3148139"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-05-21 17:10:53.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-05-22 08:43:04.000000000","tz":120},"subject":"Add Train priorities page","message":"Add Train priorities page\n\nChange-Id: I152a6eaf5f20bb1a271a8ccf3fba5cbd30893f68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/ae237f9307192759dcb685ed4ab741c12ab99bb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/ae237f9307192759dcb685ed4ab741c12ab99bb9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
