)]}'
{"id":"openstack%2Fmonasca-specs~690987","triplet_id":"openstack%2Fmonasca-specs~master~I15d90a71878eaebed02fc95f7ba052ccb201635b","project":"openstack/monasca-specs","branch":"master","hashtags":[],"change_id":"I15d90a71878eaebed02fc95f7ba052ccb201635b","subject":"Add Ussuri release cycle priorities","status":"MERGED","created":"2019-10-24 14:44:26.000000000","updated":"2019-10-31 10:09:06.000000000","submitted":"2019-10-31 10:06:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690987-1572516389856-771dd57b","meta_rev_id":"5e800396f0f66d1c791eee84de7586a40c1ea758","_number":690987,"virtual_id_number":690987,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-31 10:06:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":2,"date":"2019-10-28 09:43:57.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-10-30 15:11:29.000000000","_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":1,"date":"2019-10-30 14:56:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":1,"date":"2019-10-31 10:02:58.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 08:16:45.000000000","updated_by":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"reviewer":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"state":"REVIEWER"},{"updated":"2019-10-25 08:16:48.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2019-10-30 14:56:50.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2019-10-30 15:11:29.000000000","updated_by":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"reviewer":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"state":"REVIEWER"},{"updated":"2019-10-31 10:02:58.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-10-31 10:06:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b16046fdfd308e898b06b1f4b91e468214e7e858","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-10-24 14:44:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92850f5fb0a6eac30d704c46f5beb6a8fe02c268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 14:53:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af41f758496d498d986ea88d562d8bf9 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"6995eecee9fb82d6d7d2a5b175339c3112a25901","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2019-10-24 14:56:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"45e922b02df34e1f4661ad14396611563198eaf7","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-10-25 08:15:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"49bb0efb4b9e93161c1d0e559fbdb8da1d1594ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 08:25:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abb959ab74664bd3b2a72c7c7c1accbe : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2},{"id":"b5c17390c0a929e85399e4ae90e8bedd8e68c1ce","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2019-10-25 09:19:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8840bd34b70a0c168be0e72db71cfea3c542a6db","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-10-27 05:18:33.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nLooks good, but one small typo","accounts_in_message":[],"_revision_number":2},{"id":"1013e923c2b4c470b0217f0d96c80dd1445bc659","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-10-28 09:43:28.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"fa616fbfb293c8002d482a0d1f18735e6d8e37a5","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-10-28 09:43:57.000000000","message":"Patch Set 3: Code-Review+2\n\nLooks good to me, thanks for writing it up!","accounts_in_message":[],"_revision_number":3},{"id":"22c2d6d96e3ff13bb44a677437b39d9265f86ba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 09:49:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae78293bf5fe41a5a7b25bab830312e9 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":3},{"id":"8d6b2a68aa3ab9d625cb724288b7f182f37bf25d","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2019-10-30 14:56:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a5f02e7172d1ad2069e242741a74e402cd5b09ae","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-10-30 15:11:29.000000000","message":"Patch Set 3: Code-Review+2\n\ntypo fixed, looks good","accounts_in_message":[],"_revision_number":3},{"id":"6e9a30d8d7a5382e86c72eb7221a0d0c84128f73","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-10-31 10:02:58.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b11886b1a75bcdb8227421cdf1281fe3bc148a60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 10:03:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cfe8209c8ea37bfa4afc00f81786b2467685347a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 10:06:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"13263176080e8e5f2f29ae1aa223e9c7ff291a0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 10:06:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed4174183374ecbaed1a9f43ad32c33 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":3},{"id":"5e800396f0f66d1c791eee84de7586a40c1ea758","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 10:09:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/ac0fee0c44e3443a892ab88558d51bcd : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"399a3ed9541e4f86d08bd0823c319e0137d097ba","revisions":{"599554bbfa6f6e3beea9be26b8b61ec6f5f4c059":{"kind":"REWORK","_number":1,"created":"2019-10-24 14:44:26.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/87/690987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/87/690987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/87/690987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/87/690987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/87/690987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/87/690987/1"}}},"commit":{"parents":[{"commit":"73ab2651736c99b5cf9b8cccbc7adc0a9af06a6f","subject":"Merge \"Metrics retention policy enhancement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/73ab2651736c99b5cf9b8cccbc7adc0a9af06a6f"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-10-24 14:43:46.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-10-24 14:43:46.000000000","tz":120},"subject":"Add Ussuri release cycle priorities","message":"Add Ussuri release cycle priorities\n\nChange-Id: I15d90a71878eaebed02fc95f7ba052ccb201635b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/599554bbfa6f6e3beea9be26b8b61ec6f5f4c059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/599554bbfa6f6e3beea9be26b8b61ec6f5f4c059"}]},"branch":"refs/heads/master"},"38faf53b13ebb1e5bb6575885e8a40b4c9c85b05":{"kind":"REWORK","_number":2,"created":"2019-10-25 08:15:45.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/87/690987/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/87/690987/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/87/690987/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/87/690987/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/87/690987/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/87/690987/2"}}},"commit":{"parents":[{"commit":"73ab2651736c99b5cf9b8cccbc7adc0a9af06a6f","subject":"Merge \"Metrics retention policy enhancement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/73ab2651736c99b5cf9b8cccbc7adc0a9af06a6f"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-10-24 14:43:46.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-10-25 08:14:47.000000000","tz":120},"subject":"Add Ussuri release cycle priorities","message":"Add Ussuri release cycle priorities\n\nChange-Id: I15d90a71878eaebed02fc95f7ba052ccb201635b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/38faf53b13ebb1e5bb6575885e8a40b4c9c85b05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/38faf53b13ebb1e5bb6575885e8a40b4c9c85b05"}]},"branch":"refs/heads/master"},"399a3ed9541e4f86d08bd0823c319e0137d097ba":{"kind":"REWORK","_number":3,"created":"2019-10-28 09:43:28.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/87/690987/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/87/690987/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/87/690987/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/87/690987/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/87/690987/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/87/690987/3"}}},"commit":{"parents":[{"commit":"73ab2651736c99b5cf9b8cccbc7adc0a9af06a6f","subject":"Merge \"Metrics retention policy enhancement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/73ab2651736c99b5cf9b8cccbc7adc0a9af06a6f"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-10-24 14:43:46.000000000","tz":120},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-10-28 09:43:16.000000000","tz":0},"subject":"Add Ussuri release cycle priorities","message":"Add Ussuri release cycle priorities\n\nChange-Id: I15d90a71878eaebed02fc95f7ba052ccb201635b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/399a3ed9541e4f86d08bd0823c319e0137d097ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/399a3ed9541e4f86d08bd0823c319e0137d097ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
