)]}'
{"id":"openstack%2Fmonasca-specs~738596","triplet_id":"openstack%2Fmonasca-specs~master~Idb7514a23335392e88ea44fd0aa81159c4e2ae49","project":"openstack/monasca-specs","branch":"master","hashtags":[],"change_id":"Idb7514a23335392e88ea44fd0aa81159c4e2ae49","subject":"Add Victoria release cycle priorities","status":"MERGED","created":"2020-06-30 12:47:15.000000000","updated":"2020-07-07 13:18:35.000000000","submitted":"2020-07-07 13:15:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738596-1594127710052-abf989b5","meta_rev_id":"2bfdb531ae342fa7b839f67ec8dc9ed39c486415","_number":738596,"virtual_id_number":738596,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-07 13:15:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-07 10:57:17.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":2,"date":"2020-07-07 09:03:51.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"value":2,"date":"2020-07-07 09:21:32.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"value":1,"date":"2020-07-07 13:08:57.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-07 09:03:09.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-07-07 10:57:17.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2020-07-07 13:08:57.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2020-07-07 13:15:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb6957785a1aa6f78898cb6aa392dd0d7bfa4ce6","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-06-30 12:47:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4eeda1a972e6d75ed09e4fbe8896d18d8dec94a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 12:59:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1204df3d6a774e0395a1771f4bbe9cce : FAILURE in 6m 14s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/priorities/victoria-priorities.rst in the commit\u0027\n  Comments left for invalid file doc/source/priorities/victoria-priorities.rst","accounts_in_message":[],"_revision_number":1},{"id":"318113e76eda76add9514a5b844acc71b14c5fb1","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-07-07 08:23:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ed826cae5464e803568c6e807828a689304a6ea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 08:32:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4ce318e95ca4cdcac4bbfc555d4cb4c : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"16083f7495d0e2c0d173c7c4af533cec36f0038c","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-07-07 09:03:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fdbc842609b9caeee4a21f56d247911cb1ba0985","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-07-07 09:21:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"335a384898a7bf921590b02182607a9bba2cd87d","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2020-07-07 10:57:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"93ec96a9ea68ebf6644bbee9b711f63ba3daec7e","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-07-07 13:08:57.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fa1f4de0a6e6a7c67786b815a024a5c9213f5346","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 13:09:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b780a439a41a3fa0cbcbfcb270e4a07ad48a8fe0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 13:15:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/662a0da98f4f49c28cca64e2260dedc9 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":2},{"id":"957fedcf1aba6aa479bd063286e069ba23c2f4c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 13:15:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2bfdb531ae342fa7b839f67ec8dc9ed39c486415","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 13:18:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/7228964cfd7b497eaa80ec90358ba4ea : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a9a886f326373d1842be009091b0ee3ba8d32d42","revisions":{"588e5ec611e6d868343c94fe62c25b6ff450f17b":{"kind":"REWORK","_number":1,"created":"2020-06-30 12:47:15.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/96/738596/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/96/738596/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/96/738596/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/96/738596/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/96/738596/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/96/738596/1"}}},"commit":{"parents":[{"commit":"3c2f6f71e97e37ea9b92507fb37bc93450ce1477","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/3c2f6f71e97e37ea9b92507fb37bc93450ce1477"}]}],"author":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-06-30 12:45:30.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-06-30 12:45:30.000000000","tz":120},"subject":"Add Victoria release cycle priorities","message":"Add Victoria release cycle priorities\n\nChange-Id: Idb7514a23335392e88ea44fd0aa81159c4e2ae49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/588e5ec611e6d868343c94fe62c25b6ff450f17b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/588e5ec611e6d868343c94fe62c25b6ff450f17b"}]},"branch":"refs/heads/master"},"a9a886f326373d1842be009091b0ee3ba8d32d42":{"kind":"REWORK","_number":2,"created":"2020-07-07 08:23:00.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/96/738596/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/96/738596/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/96/738596/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/96/738596/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/96/738596/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/96/738596/2"}}},"commit":{"parents":[{"commit":"3c2f6f71e97e37ea9b92507fb37bc93450ce1477","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/3c2f6f71e97e37ea9b92507fb37bc93450ce1477"}]}],"author":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-06-30 12:45:30.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-07-07 08:22:49.000000000","tz":120},"subject":"Add Victoria release cycle priorities","message":"Add Victoria release cycle priorities\n\nChange-Id: Idb7514a23335392e88ea44fd0aa81159c4e2ae49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/a9a886f326373d1842be009091b0ee3ba8d32d42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/a9a886f326373d1842be009091b0ee3ba8d32d42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
