)]}'
{"id":"openstack%2Fmonasca-specs~769282","triplet_id":"openstack%2Fmonasca-specs~master~I88b4a8bb778efdcb2a538bc54bb18509de7b5f09","project":"openstack/monasca-specs","branch":"master","hashtags":[],"change_id":"I88b4a8bb778efdcb2a538bc54bb18509de7b5f09","subject":"remove unicode from code","status":"ABANDONED","created":"2021-01-05 09:23:09.000000000","updated":"2021-01-05 09:44:22.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"1b16b153a3965d196e21730964d86d4d32e421c5","_number":769282,"virtual_id_number":769282,"owner":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-05 09:31:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":-1,"date":"2021-01-05 09:26:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-05 09:26:25.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2021-01-05 09:31:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46a673cf3668cdb556572c4fcd044a23fc04756b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2021-01-05 09:23:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02c98e8d3c31a3799a369412c18cc289dcb937c8","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2021-01-05 09:26:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a8ff9a3f0612d3b31fb67280aef994d501fc925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 09:31:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b70aa6d801b4487fa85eda4db1909d53 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"1b16b153a3965d196e21730964d86d4d32e421c5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2021-01-05 09:44:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d08919a4ca91290c8a59fd840e63b26aea24c51d","revisions":{"d08919a4ca91290c8a59fd840e63b26aea24c51d":{"kind":"REWORK","_number":1,"created":"2021-01-05 09:23:09.000000000","uploader":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"ref":"refs/changes/82/769282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/82/769282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/82/769282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/82/769282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/82/769282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/82/769282/1"}}},"commit":{"parents":[{"commit":"a9a886f326373d1842be009091b0ee3ba8d32d42","subject":"Add Victoria release cycle priorities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/a9a886f326373d1842be009091b0ee3ba8d32d42"}]}],"author":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2021-01-05 09:22:50.000000000","tz":480},"committer":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2021-01-05 09:23:04.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: I88b4a8bb778efdcb2a538bc54bb18509de7b5f09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/d08919a4ca91290c8a59fd840e63b26aea24c51d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/d08919a4ca91290c8a59fd840e63b26aea24c51d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
