)]}'
{"id":"openstack%2Fmonasca-specs~769299","triplet_id":"openstack%2Fmonasca-specs~master~I024787954f556cf2c922c0dc31fa7fda749ce271","project":"openstack/monasca-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"32020":{"account":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"last_update":"2025-11-10 20:29:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I024787954f556cf2c922c0dc31fa7fda749ce271","subject":"remove unicode from code","status":"ABANDONED","created":"2021-01-05 09:44:31.000000000","updated":"2025-11-10 20:31:06.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"06ea6269cfdc38a198a211dd129e7919a7035738","_number":769299,"virtual_id_number":769299,"owner":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-05 09:51:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":-1,"date":"2023-02-10 01:46:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-05 09:51:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-10 01:46:48.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2025-11-10 20:31:06.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"28271f718b172832afcec4ebb7b23b0c5d406777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2021-01-05 09:44:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c307af226d16acec713672235b0458e457b66e44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 09:51:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c92c070806c24e5ea8f3ec730cd3a53e : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"4f3c827edf279e266304ad9e0097593d11e51549","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-10 01:46:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bfe25fc4bdaa0973626bfc0b3931b5cdc9f76cc7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-11-10 20:29:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"06ea6269cfdc38a198a211dd129e7919a7035738","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-11-10 20:31:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc4a655e015c22933a97938ddc465791d199e8a1","revisions":{"bc4a655e015c22933a97938ddc465791d199e8a1":{"kind":"REWORK","_number":1,"created":"2021-01-05 09:44:31.000000000","uploader":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"ref":"refs/changes/99/769299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/99/769299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/99/769299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/99/769299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/99/769299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/99/769299/1"}}},"commit":{"parents":[{"commit":"a9a886f326373d1842be009091b0ee3ba8d32d42","subject":"Add Victoria release cycle priorities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/a9a886f326373d1842be009091b0ee3ba8d32d42"}]}],"author":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2021-01-05 09:44:09.000000000","tz":480},"committer":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2021-01-05 09:44:26.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: I024787954f556cf2c922c0dc31fa7fda749ce271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/bc4a655e015c22933a97938ddc465791d199e8a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/bc4a655e015c22933a97938ddc465791d199e8a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
