)]}'
{"id":"openstack%2Fmonasca-specs~855040","triplet_id":"openstack%2Fmonasca-specs~master~I6e87f918c5b2ba47d47c0632333e80160bc84c94","project":"openstack/monasca-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"28062":{"account":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"last_update":"2026-02-04 05:29:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6e87f918c5b2ba47d47c0632333e80160bc84c94","subject":"Add Zed release cycle priorities","status":"ABANDONED","created":"2022-08-29 15:01:16.000000000","updated":"2026-02-04 05:29:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"811808485f5acac8c99b7b4d87f6bdd6b64ca3ec","_number":855040,"virtual_id_number":855040,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-29 15:08:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-29 15:08:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-04 05:29:13.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"108f4ea4f31acc5c88d431445c15430aee625e3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2022-08-29 15:01:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1cfe9fbb8d3a6645f9ddfd571d5d4be8062ca62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:08:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c0a2a35a9cc410a94cd49f9e3e53eac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/471770b0ea9d4d8190d5cb348fc57485 : FAILURE in 2m 22s","accounts_in_message":[],"_revision_number":1},{"id":"811808485f5acac8c99b7b4d87f6bdd6b64ca3ec","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-04 05:29:13.000000000","message":"Abandoned\n\nThis project has been retired and no longer accepts code changes via https://review.opendev.org. New changes can be uploaded if the project is restored.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ea0b4d073bae81af6a1c03eee4c4ae0644fa427","revisions":{"6ea0b4d073bae81af6a1c03eee4c4ae0644fa427":{"kind":"REWORK","_number":1,"created":"2022-08-29 15:01:16.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/40/855040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-specs","ref":"refs/changes/40/855040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/40/855040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/40/855040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-specs refs/changes/40/855040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-specs refs/changes/40/855040/1"}}},"commit":{"parents":[{"commit":"393acda2c9eecdab55e43f33f250df996bffcc67","subject":"Add Yoga release cycle priorities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/393acda2c9eecdab55e43f33f250df996bffcc67"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2022-08-29 15:00:15.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2022-08-29 15:00:15.000000000","tz":120},"subject":"Add Zed release cycle priorities","message":"Add Zed release cycle priorities\n\nChange-Id: I6e87f918c5b2ba47d47c0632333e80160bc84c94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/6ea0b4d073bae81af6a1c03eee4c4ae0644fa427"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-specs/commit/6ea0b4d073bae81af6a1c03eee4c4ae0644fa427"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
