)]}'
{"id":"openstack%2Fmonasca-statsd~670122","triplet_id":"openstack%2Fmonasca-statsd~master~I4f2855d6090b91f413f5bfc20b6183ef4ded7e42","project":"openstack/monasca-statsd","branch":"master","hashtags":[],"change_id":"I4f2855d6090b91f413f5bfc20b6183ef4ded7e42","subject":"Disable monasca-tempest-java-influxdb job","status":"MERGED","created":"2019-07-10 16:27:02.000000000","updated":"2019-07-15 08:56:16.000000000","submitted":"2019-07-15 08:56:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670122-1563180976220-3eb45556","meta_rev_id":"6e529b7166acc8211f17cbd59f55bfe5de3dff4f","_number":670122,"virtual_id_number":670122,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-15 08:56:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-15 08:24:45.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-15 08:25:04.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-15 08:25:04.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2019-07-15 08:56:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"980f8a92363e58b7f65ec1987116cd1432d977e3","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-10 16:27:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"300894c94a7cda4bd226fc69ce1c5c6ba3bc7ea1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:21:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/22/670122/1/check/openstack-tox-cover/a55f138/cover/ : SUCCESS in 3m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/22/670122/1/check/openstack-tox-lower-constraints/6e77ba9/ : SUCCESS in 4m 51s\n- openstack-tox-pep8 http://logs.openstack.org/22/670122/1/check/openstack-tox-pep8/f8014d5/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/22/670122/1/check/openstack-tox-py27/1c264a6/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/22/670122/1/check/openstack-tox-py36/0f68129/ : SUCCESS in 3m 37s\n- openstack-tox-py37 http://logs.openstack.org/22/670122/1/check/openstack-tox-py37/20250be/ : SUCCESS in 3m 19s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/22/670122/1/check/monasca-tempest-python2-influxdb/45459b1/ : SUCCESS in 32m 46s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/22/670122/1/check/monasca-tempest-python3-influxdb/69ab85a/ : FAILURE in 43m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7a1a0f3aeda7c30065dd3000553360267461a1e","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-07-15 08:24:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0901f90f9e0d5a0e517fadebbbd34d5a9529e407","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-07-15 08:25:04.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e3d5a5800e2a6e44ea52467c73890d5ae0c03a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 08:25:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2a3d50f96984d812ff059655684e3b401fd9997","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 08:56:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6e529b7166acc8211f17cbd59f55bfe5de3dff4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 08:56:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/22/670122/1/gate/openstack-tox-lower-constraints/8fd1d15/ : SUCCESS in 5m 10s\n- openstack-tox-pep8 http://logs.openstack.org/22/670122/1/gate/openstack-tox-pep8/a53ed21/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/22/670122/1/gate/openstack-tox-py27/b7b4f16/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/22/670122/1/gate/openstack-tox-py36/7f2d214/ : SUCCESS in 3m 48s\n- openstack-tox-py37 http://logs.openstack.org/22/670122/1/gate/openstack-tox-py37/185c222/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a8ce64b366fa198303fa1cac2292472e64b56dc","revisions":{"6a8ce64b366fa198303fa1cac2292472e64b56dc":{"kind":"REWORK","_number":1,"created":"2019-07-10 16:27:02.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/22/670122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-statsd","ref":"refs/changes/22/670122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/22/670122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/22/670122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/22/670122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-statsd refs/changes/22/670122/1"}}},"commit":{"parents":[{"commit":"b9bdf0c7926f268a6c8134b351f0880edf05a4f0","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/b9bdf0c7926f268a6c8134b351f0880edf05a4f0"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-07-10 16:26:13.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-07-10 16:26:13.000000000","tz":120},"subject":"Disable monasca-tempest-java-influxdb job","message":"Disable monasca-tempest-java-influxdb job\n\nThe job has been failing for a long time. Java implementation of\nInfluxDB persister is not maintained.\n\nChange-Id: I4f2855d6090b91f413f5bfc20b6183ef4ded7e42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/6a8ce64b366fa198303fa1cac2292472e64b56dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/6a8ce64b366fa198303fa1cac2292472e64b56dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
