)]}'
{"id":"openstack%2Fmonasca-statsd~690501","triplet_id":"openstack%2Fmonasca-statsd~master~I240f95772c990dc1ee84ba5ece50573dd2079240","project":"openstack/monasca-statsd","branch":"master","topic":"ussuri","hashtags":[],"change_id":"I240f95772c990dc1ee84ba5ece50573dd2079240","subject":"Switch to Ussuri jobs","status":"MERGED","created":"2019-10-23 08:09:47.000000000","updated":"2019-10-23 09:40:45.000000000","submitted":"2019-10-23 09:40:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690501-1571823645432-9e4bfd77","meta_rev_id":"b7387f8862060ef93410313e8afc973400da6b7c","_number":690501,"virtual_id_number":690501,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 09:40:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-23 09:29:24.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-23 09:29:24.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 09:29:24.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-10-23 09:40:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bea9e698736031725492cf3078e06a9f8bc8c9e","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-23 08:09:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6a38fa25d35e0bc79a270f19e3320c9acbd36b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 08:56:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cca63aa15c994080a872486d957ad36b : SUCCESS in 3m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb91c64004c04b719e2551076298fd0b : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee35e7e0c294434887ff38d630ca26ec : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75087ec3dbaf465d9de5ae225a1f009f : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13011ab437ef43309eea39a07266cfa1 : SUCCESS in 2m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/510377a76a354104ad359a1e0a0a210c : SUCCESS in 5m 40s\n- monasca-tempest-python2-influxdb https://zuul.opendev.org/t/openstack/build/e0eafd176874454689710a62b9770498 : SUCCESS in 38m 53s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/cb686e6383f64a13a8db3cedf8193c1c : SUCCESS in 39m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e01a393da9aca4c8b29b33e073cbf232e239a1f","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-10-23 09:29:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db1b977bfc30aa1f77b26930104354cd55ed2fbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 09:29:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29964c701b3ff0741608737a249cd56894ceacb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 09:40:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"88e4eaed7a8dcd20c832ca52edeccfb611c86641","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 09:40:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a3b05200e524e87bcf31e7cc0980c85 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca1d4ba18bf94d5b9429afc9f5a87328 : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b50874cca171437f9ebffc93c8f8d91f : SUCCESS in 2m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1245176bcff045568e31f0e0916aec3a : SUCCESS in 2m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d806fc2c6fa4876ae2cfdd233f9cb37 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"680d9118686d5234f8b9196f81cff40b7a168f9b","revisions":{"680d9118686d5234f8b9196f81cff40b7a168f9b":{"kind":"REWORK","_number":1,"created":"2019-10-23 08:09:47.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/01/690501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-statsd","ref":"refs/changes/01/690501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/01/690501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/01/690501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/01/690501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-statsd refs/changes/01/690501/1"}}},"commit":{"parents":[{"commit":"5d2426f88e06db7e80526b32b6955edc256a629b","subject":"Update the constraints url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/5d2426f88e06db7e80526b32b6955edc256a629b"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-23 08:10:11.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-23 08:10:11.000000000","tz":480},"subject":"Switch to Ussuri jobs","message":"Switch to Ussuri jobs\n\nChange-Id: I240f95772c990dc1ee84ba5ece50573dd2079240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/680d9118686d5234f8b9196f81cff40b7a168f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/680d9118686d5234f8b9196f81cff40b7a168f9b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
