)]}'
{"id":"openstack%2Fmonasca-statsd~723919","triplet_id":"openstack%2Fmonasca-statsd~master~I43a802996b1d26389e4a59ec170e246646de86b8","project":"openstack/monasca-statsd","branch":"master","topic":"add-victoria-python-jobtemplates","hashtags":[],"change_id":"I43a802996b1d26389e4a59ec170e246646de86b8","subject":"Add Python3 victoria unit tests","status":"MERGED","created":"2020-04-28 14:36:14.000000000","updated":"2020-04-28 15:52:28.000000000","submitted":"2020-04-28 15:52:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723919-1588089148732-bde3a662","meta_rev_id":"ae99b54e666ae52ff9ae395ced53e78087620371","_number":723919,"virtual_id_number":723919,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-28 15:52:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-28 15:35:02.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-28 15:35:02.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 15:35:02.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2020-04-28 15:52:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4283001f8126ba0ddea7728a88014dc73c81fe55","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-28 14:36:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff12807b190b3042fc2aaf460062ead1a7452674","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 15:27:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/668819b97fac49acaa6cadedfe4197e0 : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f38d961905844cea8bafacf6f4f3f8da : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95165bfd10d84433953c31cf9ae62821 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2926df730054948aa0b708d1f0278de : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bb11cbbaa604dfcad4abfb0e3945ba1 : SUCCESS in 4m 04s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/b9bf579128b049048cf31660d5bfbac8 : SUCCESS in 38m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a882ab3c2b5852dc9b0ddff245b839402baab7a4","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-04-28 15:35:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f4b5d405045c710485b5d4db6142b77f73438472","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 15:35:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bcf8e2b46ceec447d7b2e334146b236a8f19f8f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 15:52:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8f7ac000bb184c0cb5134a3ce7e25c37f607fd57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 15:52:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d23cad0405e4952886a5fecc5ed27a3 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7486b681a4d34c29b8aea5a883d7c072 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ece3196c49b8469cbfb7b6020443d4af : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2732372f1d74671adbae18014b051a3 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6e36cb0afe383c7d8973bd418a1b579bc7c3935","revisions":{"d6e36cb0afe383c7d8973bd418a1b579bc7c3935":{"kind":"REWORK","_number":1,"created":"2020-04-28 14:36:14.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/19/723919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-statsd","ref":"refs/changes/19/723919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/19/723919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/19/723919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/19/723919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-statsd refs/changes/19/723919/1"}}},"commit":{"parents":[{"commit":"257d6e42b07e3ebd4598f2576d8d212790cf7bb7","subject":"Merge \"Update hacking for Python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/257d6e42b07e3ebd4598f2576d8d212790cf7bb7"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:36:12.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:36:12.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I43a802996b1d26389e4a59ec170e246646de86b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/d6e36cb0afe383c7d8973bd418a1b579bc7c3935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/d6e36cb0afe383c7d8973bd418a1b579bc7c3935"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
