)]}'
{"id":"openstack%2Fmonasca-statsd~731274","triplet_id":"openstack%2Fmonasca-statsd~master~Ie82e6bf3964c84413e6af0c94e1212b7669de93b","project":"openstack/monasca-statsd","branch":"master","topic":"stestr","hashtags":[],"change_id":"Ie82e6bf3964c84413e6af0c94e1212b7669de93b","subject":"Fix and unify cover environment","status":"MERGED","created":"2020-05-27 16:38:06.000000000","updated":"2020-05-28 12:13:04.000000000","submitted":"2020-05-28 12:13:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731274-1590667984666-3a602513","meta_rev_id":"2b8aa0144f5168a23ebe50486a68b84ba6f51ed3","_number":731274,"virtual_id_number":731274,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 12:13:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-28 11:59:18.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-28 11:59:18.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 11:59:18.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2020-05-28 12:13:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2065cd9bd3a8049f7b1d54777286799e52537e0e","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-05-27 16:38:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e594788b89dc60effa73b06d7aad326125b90614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 17:22:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5018beeccc94454dbe88da605147d183 : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91bb75dc944343d683e246f2b2702a66 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/371dee5628274eb28c0cbfa95a413a85 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61c2a656c27246baa3076982f2ef0477 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/527bdf07608b4ebab8d1609a0c446163 : SUCCESS in 5m 32s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/a074eea35c6e498085724c88a043ce80 : SUCCESS in 37m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6af7427dad3910cfa299632dc198690a5b703094","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-05-28 11:59:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"34da785bf177a6cc9347addafd4d8459e89eb3d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 11:59:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b726b4fe343c70df9b599840a1741239693f36ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 12:13:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"08363a97df9365d9ee3226ecf8049b5fe11a430b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 12:13:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/145cc6bc3e7e4eaf8b5202e4aa9bbb1f : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/954a7f06a05642cbac666283de2567ac : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76f5828b4af147788c0ddad6e14019ad : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a896447a2e4440e38d0f6cffa63781b0 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fcb40ee791c987d6090f3a760a2b6271b7f0d855","revisions":{"fcb40ee791c987d6090f3a760a2b6271b7f0d855":{"kind":"REWORK","_number":1,"created":"2020-05-27 16:38:06.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/74/731274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-statsd","ref":"refs/changes/74/731274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/74/731274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/74/731274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/74/731274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-statsd refs/changes/74/731274/1"}}},"commit":{"parents":[{"commit":"460431085eef839ee33b240645abff2271db605b","subject":"Add CONTRIBUTING.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/460431085eef839ee33b240645abff2271db605b"}]}],"author":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-05-27 16:37:16.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-05-27 16:38:04.000000000","tz":120},"subject":"Fix and unify cover environment","message":"Fix and unify cover environment\n\nFix cover environment and unify it accordingly to Project Testing\nInterface [1].\n\n[1] https://governance.openstack.org/tc/reference/pti/python.html#coverage-jobs\n\nChange-Id: Ie82e6bf3964c84413e6af0c94e1212b7669de93b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/fcb40ee791c987d6090f3a760a2b6271b7f0d855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/fcb40ee791c987d6090f3a760a2b6271b7f0d855"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
