)]}'
{"id":"openstack%2Fmonasca-statsd~769725","triplet_id":"openstack%2Fmonasca-statsd~master~Ic771f9dfa0ee59c05f986bdbf24255dda87cc977","project":"openstack/monasca-statsd","branch":"master","topic":"fix-relmgt-pip-doc","hashtags":[],"change_id":"Ic771f9dfa0ee59c05f986bdbf24255dda87cc977","subject":"Add doc/requirements","status":"MERGED","created":"2021-01-07 13:23:21.000000000","updated":"2021-01-18 16:19:51.000000000","submitted":"2021-01-18 16:19:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769725-fix-relmgt-pip-doc","meta_rev_id":"2812da7014cf6f442446c90030e5f272eb09402a","_number":769725,"virtual_id_number":769725,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-18 16:19:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-18 16:06:35.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-18 16:06:35.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-07 14:05:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-12 12:38:17.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"}],"messages":[{"id":"bd58ab93bf903090e4ab47451b88b62d25041ef0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-07 13:23:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0761b101a1dcc0598a832a86dd687abf94e24dda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 14:05:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/add6f0f42ea446b584da08379c47336f : SUCCESS in 2m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f304a471fdc4260b3181f90cb645636 : SUCCESS in 4m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8c420d546094595acddcb338892e36e : FAILURE in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/615a2577def24ea0915e622fa4ff47dc : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f125a26877884201ac15c31e664e0224 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ee691f2d5124b7fbd92a408158abfe2 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3440ef1aadd426fbf8456e79ea36276 : SUCCESS in 4m 11s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/fd26f13fa15a4f91a05023ea831a9672 : SUCCESS in 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed38c07a1123723ded5dde45bd4d9b07dc327956","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-01-18 14:58:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f9e7740a662587cea208db0d073ad5d0314710d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-01-18 15:20:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"da8246cc9db85d288816034900beb36a3396c189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:05:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0e4a7a46835d40d89db6bdb2c69bbc32 : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95c7889603934a27949e33f30b6a60b2 : SUCCESS in 4m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7da9770f907645359fb007a7f641fd91 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4afedc9575314ee4b143cada3ffe2ab8 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64782e2d76cc40a7be1ac788a00f7c65 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63c8f5b4c93347d2bb2af4b7966555f5 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/650515d6b9314f2bbc0af0fc57cfbd3b : SUCCESS in 4m 11s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/aa613972f4e04c069d320855c14ae1b0 : SUCCESS in 38m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"057000dceee1c9c4859600ae45270b74f1898ed4","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-01-18 16:06:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a9173fae3453b729f02754109342a368f8c504aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:09:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e6983e4a2cda0f1ac12a7625f58f1626849f4369","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:19:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/aa0f34e4ead54acca508ae52f84dd57c : SUCCESS in 2m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5355b20faa774d43808a2414271fd332 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9289641f2573464d8f567ee83f392255 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8a57628bcfb4f699ccecd56a818df30 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f081805ac744f67be2355197f21c436 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":3},{"id":"2812da7014cf6f442446c90030e5f272eb09402a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:19:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ed2c0d24f7bb4bf1e91beb7ff815bc008868efe7","revisions":{"e44d813a0c817a3219492cd30e78b65f37899674":{"kind":"REWORK","_number":1,"created":"2021-01-07 13:23:21.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/25/769725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-statsd","ref":"refs/changes/25/769725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/25/769725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/25/769725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/25/769725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-statsd refs/changes/25/769725/1"}}},"commit":{"parents":[{"commit":"298394511e160a436849773e1ab2c080ded231e7","subject":"Merge \"bump py37 to py38 in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/298394511e160a436849773e1ab2c080ded231e7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-07 13:22:49.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-07 13:22:49.000000000","tz":60},"subject":"Add doc/requirements","message":"Add doc/requirements\n\nWe need to specify doc requirements in doc/requirements.txt\nto avoid problems with the pip resolver [1] for the release team [2][3].\nRemoved specific doc requirements from test-requirements.txt.\n\nThe problem here is that this repos haven\u0027t doc/requirements.txt file\nand by default in this case zuul will use the test-requirements.txt file\nto pull requirements [4].\n\nThis requirements file contains extra requirements like flake8 that\ncollided with those allowed in our job environment and so the new pip\nresolver fails to install these requirements and the job exits in error.\n\nThis project meet the conditions leading to the bug.\n\n[1] http://lists.openstack.org/pipermail/release-job-failures/2021-January/001500.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019611.html\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019612.html\n[4] https://opendev.org/zuul/zuul-jobs/src/branch/master/roles/ensure-sphinx/tasks/main.yaml#L36\n\nChange-Id: Ic771f9dfa0ee59c05f986bdbf24255dda87cc977\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/e44d813a0c817a3219492cd30e78b65f37899674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/e44d813a0c817a3219492cd30e78b65f37899674"}]},"branch":"refs/heads/master"},"70126289f8f86f5f50f96cd702ad0cb1797c01f2":{"kind":"REWORK","_number":2,"created":"2021-01-18 14:58:54.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/25/769725/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-statsd","ref":"refs/changes/25/769725/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/25/769725/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/25/769725/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/25/769725/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-statsd refs/changes/25/769725/2"}}},"commit":{"parents":[{"commit":"298394511e160a436849773e1ab2c080ded231e7","subject":"Merge \"bump py37 to py38 in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/298394511e160a436849773e1ab2c080ded231e7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-07 13:22:49.000000000","tz":60},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-01-18 14:58:31.000000000","tz":60},"subject":"Add doc/requirements","message":"Add doc/requirements\n\nWe need to specify doc requirements in doc/requirements.txt\nto avoid problems with the pip resolver [1] for the release team [2][3].\nRemoved specific doc requirements from test-requirements.txt.\n\nThe problem here is that this repos haven\u0027t doc/requirements.txt file\nand by default in this case zuul will use the test-requirements.txt file\nto pull requirements [4].\n\nThis requirements file contains extra requirements like flake8 that\ncollided with those allowed in our job environment and so the new pip\nresolver fails to install these requirements and the job exits in error.\n\nThis project meet the conditions leading to the bug.\n\n[1] http://lists.openstack.org/pipermail/release-job-failures/2021-January/001500.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019611.html\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019612.html\n[4] https://opendev.org/zuul/zuul-jobs/src/branch/master/roles/ensure-sphinx/tasks/main.yaml#L36\n\nChange-Id: Ic771f9dfa0ee59c05f986bdbf24255dda87cc977\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/70126289f8f86f5f50f96cd702ad0cb1797c01f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/70126289f8f86f5f50f96cd702ad0cb1797c01f2"}]},"branch":"refs/heads/master"},"ed2c0d24f7bb4bf1e91beb7ff815bc008868efe7":{"kind":"REWORK","_number":3,"created":"2021-01-18 15:20:14.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/25/769725/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-statsd","ref":"refs/changes/25/769725/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/25/769725/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/25/769725/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-statsd refs/changes/25/769725/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-statsd refs/changes/25/769725/3"}}},"commit":{"parents":[{"commit":"298394511e160a436849773e1ab2c080ded231e7","subject":"Merge \"bump py37 to py38 in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/298394511e160a436849773e1ab2c080ded231e7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-07 13:22:49.000000000","tz":60},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-01-18 15:19:51.000000000","tz":60},"subject":"Add doc/requirements","message":"Add doc/requirements\n\nWe need to specify doc requirements in doc/requirements.txt\nto avoid problems with the pip resolver [1] for the release team [2][3].\nRemoved specific doc requirements from test-requirements.txt.\n\nThe problem here is that this repos haven\u0027t doc/requirements.txt file\nand by default in this case zuul will use the test-requirements.txt file\nto pull requirements [4].\n\nThis requirements file contains extra requirements like flake8 that\ncollided with those allowed in our job environment and so the new pip\nresolver fails to install these requirements and the job exits in error.\n\nThis project meet the conditions leading to the bug.\n\n[1] http://lists.openstack.org/pipermail/release-job-failures/2021-January/001500.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019611.html\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019612.html\n[4] https://opendev.org/zuul/zuul-jobs/src/branch/master/roles/ensure-sphinx/tasks/main.yaml#L36\n\nChange-Id: Ic771f9dfa0ee59c05f986bdbf24255dda87cc977\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/ed2c0d24f7bb4bf1e91beb7ff815bc008868efe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-statsd/commit/ed2c0d24f7bb4bf1e91beb7ff815bc008868efe7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
