)]}'
{"id":"openstack%2Fmonasca-tempest-plugin~663927","triplet_id":"openstack%2Fmonasca-tempest-plugin~master~I87c38faf41f7c1bff72d45d238a6f38751fad483","project":"openstack/monasca-tempest-plugin","branch":"master","topic":"encoding","hashtags":[],"change_id":"I87c38faf41f7c1bff72d45d238a6f38751fad483","subject":"Use unicode literals in test_metrics","status":"MERGED","created":"2019-06-07 12:51:39.000000000","updated":"2019-07-10 18:54:34.000000000","submitted":"2019-07-10 18:54:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663927-1562784874324-42ebc4e7","meta_rev_id":"53dcdc4f3756f58013f3699ac0d24c2bbfbe0e11","_number":663927,"virtual_id_number":663927,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":23735,"name":"sumitjami","email":"sjamgade@suse.com","username":"nthforloop"},{"value":2,"date":"2019-07-10 18:54:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"date":"2019-07-10 16:58:51.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"recommended":{"_account_id":23735,"name":"sumitjami","email":"sjamgade@suse.com","username":"nthforloop"},"all":[{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":1,"date":"2019-07-05 08:38:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23735,"name":"sumitjami","email":"sjamgade@suse.com","username":"nthforloop"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-05 14:23:35.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":1,"date":"2019-07-10 08:06:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"all":[{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":1,"date":"2019-07-10 11:24:42.000000000","_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"date":"2019-07-04 13:49:22.000000000","_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":23735,"name":"sumitjami","email":"sjamgade@suse.com","username":"nthforloop"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23735,"name":"sumitjami","email":"sjamgade@suse.com","username":"nthforloop"},{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 09:59:20.000000000","updated_by":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"reviewer":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"state":"REVIEWER"},{"updated":"2019-07-04 13:49:22.000000000","updated_by":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"reviewer":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"state":"REVIEWER"},{"updated":"2019-07-05 08:38:36.000000000","updated_by":{"_account_id":23735,"name":"sumitjami","email":"sjamgade@suse.com","username":"nthforloop"},"reviewer":{"_account_id":23735,"name":"sumitjami","email":"sjamgade@suse.com","username":"nthforloop"},"state":"REVIEWER"},{"updated":"2019-07-05 14:23:35.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2019-07-10 08:06:11.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2019-07-10 11:24:42.000000000","updated_by":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"reviewer":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"state":"REVIEWER"},{"updated":"2019-07-10 18:54:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"275c65ddd2d45e5f3f93b725f0183e612e7d9114","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-06-07 12:51:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1989b19bf7ed45afe95ae4c4fbf58b7b8f6cb55a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:55:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/1/check/openstack-tox-pep8/ce0f274/ : SUCCESS in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/27/663927/1/check/openstack-tox-docs/d7bd6a6/html/ : SUCCESS in 4m 08s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/1/check/monasca-tempest-python2-influxdb/28ccaf9/ : SUCCESS in 40m 23s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/1/check/monasca-tempest-python3-influxdb/8500bdc/ : FAILURE in 43m 59s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/1/check/monasca-tempest-python2-cassandra/d6f8922/ : SUCCESS in 35m 29s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/1/check/monascalog-python2-tempest/1e78145/ : FAILURE in 38m 45s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/1/check/monascalog-python3-tempest/f0e66a4/ : FAILURE in 43m 02s\n- build-monasca-docker-image http://logs.openstack.org/27/663927/1/check/build-monasca-docker-image/68e70b6/ : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":1},{"id":"3dbabdba0f6e27426a78adcb616b307948f55251","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-06-07 15:24:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"499bd18c484a1e8cf0bf77b80f6eb5df94f3fd50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 16:26:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/1/check/openstack-tox-pep8/2347173/ : SUCCESS in 4m 33s\n- openstack-tox-docs http://logs.openstack.org/27/663927/1/check/openstack-tox-docs/59a41f4/html/ : SUCCESS in 4m 29s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/1/check/monasca-tempest-python2-influxdb/dd7c924/ : SUCCESS in 39m 47s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/1/check/monasca-tempest-python3-influxdb/8a52e58/ : FAILURE in 45m 34s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/1/check/monasca-tempest-python2-cassandra/b44b4ba/ : SUCCESS in 45m 26s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/1/check/monascalog-python2-tempest/1a352be/ : FAILURE in 47m 36s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/1/check/monascalog-python3-tempest/e3fdd6a/ : FAILURE in 47m 07s\n- build-monasca-docker-image http://logs.openstack.org/27/663927/1/check/build-monasca-docker-image/73104f6/ : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b170bc841596b5c9ab4f9ead98cc184a3ae39005","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-06-20 09:59:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d41390e4b4efaf8c249246cdf30628c26c0708a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 12:03:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/1/check/openstack-tox-pep8/7dc4c17/ : SUCCESS in 4m 37s\n- openstack-tox-docs http://logs.openstack.org/27/663927/1/check/openstack-tox-docs/f28c2be/html/ : SUCCESS in 4m 01s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/1/check/monasca-tempest-python2-influxdb/93570f3/ : SUCCESS in 37m 17s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/1/check/monasca-tempest-python3-influxdb/74dfe29/ : FAILURE in 41m 53s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/1/check/monasca-tempest-python2-cassandra/3aa16bc/ : SUCCESS in 42m 44s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/1/check/monascalog-python2-tempest/2b3e4d9/ : SUCCESS in 40m 39s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/1/check/monascalog-python3-tempest/661578b/ : SUCCESS in 1h 53m 09s\n- build-monasca-docker-image http://logs.openstack.org/27/663927/1/check/build-monasca-docker-image/8ec0af2/ : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":1},{"id":"bc657b566b02641e4ec83ddc0234686413489499","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-06-24 08:51:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b1d1e54e8d6dcc662ab579ef7da23f95bd13ffea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 10:01:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/2/check/openstack-tox-pep8/bca5c12/ : SUCCESS in 4m 38s\n- openstack-tox-docs http://logs.openstack.org/27/663927/2/check/openstack-tox-docs/5e95964/html/ : SUCCESS in 4m 16s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/2/check/monasca-tempest-python2-influxdb/533b2bf/ : SUCCESS in 40m 52s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/2/check/monasca-tempest-python3-influxdb/62ef014/ : FAILURE in 39m 12s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/2/check/monasca-tempest-python2-cassandra/06e51aa/ : SUCCESS in 35m 36s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/2/check/monascalog-python2-tempest/1a37cb3/ : SUCCESS in 42m 27s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/2/check/monascalog-python3-tempest/d3352b9/ : SUCCESS in 41m 28s\n- build-monasca-docker-image http://logs.openstack.org/27/663927/2/check/build-monasca-docker-image/903b23e/ : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":2},{"id":"482c1fe8a5e1bae4278d6b78529ed76ddc2fa820","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-06-26 12:08:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3697a9615bcf1f9fee5509cc265e25234b6fccd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 13:06:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/3/check/openstack-tox-pep8/23f2e94/ : SUCCESS in 3m 13s\n- openstack-tox-docs http://logs.openstack.org/27/663927/3/check/openstack-tox-docs/5fd2bde/html/ : SUCCESS in 3m 37s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/3/check/monasca-tempest-python2-influxdb/5d23564/ : FAILURE in 45m 02s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/3/check/monasca-tempest-python3-influxdb/707cfd7/ : FAILURE in 51m 39s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/3/check/monasca-tempest-python2-cassandra/bdea494/ : SUCCESS in 39m 17s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/3/check/monascalog-python2-tempest/411d42f/ : SUCCESS in 48m 41s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/3/check/monascalog-python3-tempest/b7300f1/ : SUCCESS in 49m 30s\n- build-monasca-docker-image http://logs.openstack.org/27/663927/3/check/build-monasca-docker-image/dbc038d/ : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":3},{"id":"00a4f0c6c8a3faf09d7b04c99d82db336e43aa5b","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-07-01 13:03:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c11ac89f6c37609ac97dc191e880887ad5adb69a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 13:58:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/3/check/openstack-tox-pep8/3263c98/ : SUCCESS in 3m 47s\n- openstack-tox-docs http://logs.openstack.org/27/663927/3/check/openstack-tox-docs/e39afcb/html/ : SUCCESS in 3m 26s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/3/check/monasca-tempest-python2-influxdb/a805584/ : SUCCESS in 37m 15s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/3/check/monasca-tempest-python3-influxdb/6de5e31/ : FAILURE in 45m 22s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/3/check/monasca-tempest-python2-cassandra/c065bc6/ : SUCCESS in 39m 31s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/3/check/monascalog-python2-tempest/7e44185/ : SUCCESS in 39m 01s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/3/check/monascalog-python3-tempest/f323a06/ : SUCCESS in 47m 32s\n- build-monasca-docker-image http://logs.openstack.org/27/663927/3/check/build-monasca-docker-image/e9660f6/ : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":3},{"id":"39ad163f2b245521e2f35bf2d75524ac2d3d2bb1","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2019-07-02 12:19:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d7b0726cc82b70cff8588c907e85ff5c9aec8554","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 13:05:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/4/check/openstack-tox-pep8/ba84e33/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/27/663927/4/check/openstack-tox-docs/dc9268b/html/ : SUCCESS in 2m 46s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/4/check/monasca-tempest-python2-influxdb/4bad839/ : SUCCESS in 38m 43s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/4/check/monasca-tempest-python3-influxdb/a157cc9/ : SUCCESS in 36m 55s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/4/check/monasca-tempest-python2-cassandra/560f9cb/ : SUCCESS in 33m 32s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/4/check/monascalog-python2-tempest/2d836e7/ : SUCCESS in 39m 08s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/4/check/monascalog-python3-tempest/3089bbe/ : SUCCESS in 40m 23s\n- build-monasca-docker-image http://logs.openstack.org/27/663927/4/check/build-monasca-docker-image/be782d8/ : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":4},{"id":"2cfd433503d404afbb21955b24f85a757a0eabb2","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-02 15:59:25.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"22c043279f197f312ec8493d6277422759852893","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-03 08:43:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1b459f60231a1966af9bd41be1c10b3ce627e473","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-03 08:57:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2b785ab447a8e0a9edce551df15296217d07fc47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 09:53:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/6/check/openstack-tox-pep8/6fd39c6/ : SUCCESS in 3m 03s\n- openstack-tox-docs http://logs.openstack.org/27/663927/6/check/openstack-tox-docs/67e9711/html/ : SUCCESS in 3m 34s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/6/check/monasca-tempest-python2-influxdb/2a260e8/ : SUCCESS in 31m 33s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/6/check/monasca-tempest-python3-influxdb/3cd0aee/ : FAILURE in 44m 23s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/6/check/monasca-tempest-python2-cassandra/8b8ff21/ : FAILURE in 13m 36s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/6/check/monascalog-python2-tempest/3e0c026/ : SUCCESS in 48m 04s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/6/check/monascalog-python3-tempest/8509b41/ : SUCCESS in 48m 52s\n- build-monasca-docker-image http://logs.openstack.org/27/663927/6/check/build-monasca-docker-image/7e24a23/ : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":6},{"id":"43d8ef839247c9059b96df00e2bd937e0c3c9177","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-03 09:57:35.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"1e3a51bb52db841be6c33d2e53c1418ed5f3a2f5","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-03 10:40:09.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"f8c21e38ac54e18be82bba855fd3471e6820f17a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 11:03:25.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/6/check/openstack-tox-pep8/7df6c13/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/27/663927/6/check/openstack-tox-docs/544b0a7/ : POST_FAILURE in 17m 00s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/6/check/monasca-tempest-python2-influxdb/0288ff5/ : SUCCESS in 39m 49s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/6/check/monasca-tempest-python3-influxdb/b87236c/ : FAILURE in 27m 15s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/6/check/monasca-tempest-python2-cassandra/c89eb53/ : SUCCESS in 39m 30s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/6/check/monascalog-python2-tempest/cf66f93/ : SUCCESS in 38m 14s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/6/check/monascalog-python3-tempest/5ae02dd/ : SUCCESS in 57m 35s\n- build-monasca-docker-image http://logs.openstack.org/27/663927/6/check/build-monasca-docker-image/43bd781/ : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":6},{"id":"52c77a73b41c175d4f320a7d3706fd8d5b5be47b","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-03 11:16:58.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"5509f3f374622f1ea07c84e187289f8be87fc762","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 12:10:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/6/check/openstack-tox-pep8/54e0290/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/27/663927/6/check/openstack-tox-docs/64c0c1b/html/ : SUCCESS in 3m 14s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/6/check/monasca-tempest-python2-influxdb/b0e2200/ : SUCCESS in 37m 23s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/6/check/monasca-tempest-python3-influxdb/9b03218/ : SUCCESS in 39m 07s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/6/check/monasca-tempest-python2-cassandra/30f1abb/ : SUCCESS in 40m 35s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/6/check/monascalog-python2-tempest/b62441c/ : SUCCESS in 39m 11s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/6/check/monascalog-python3-tempest/bba7de8/ : SUCCESS in 49m 07s\n- build-monasca-docker-image http://logs.openstack.org/27/663927/6/check/build-monasca-docker-image/d542487/ : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":6},{"id":"e2bfea64a265219c8bbd08c223e07d5fe01a19c8","author":{"_account_id":23735,"name":"sumitjami","email":"sjamgade@suse.com","username":"nthforloop"},"date":"2019-07-05 08:38:36.000000000","message":"Patch Set 6: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"232e01afceb496f2e731c9e086966fd74be278a1","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-07-05 14:23:35.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"c95506b415fd4c260f9d8b92c716eb8874003a48","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2019-07-10 08:06:11.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"12b583c5db6931d08928b173d05bde4a24b72829","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-07-10 11:24:42.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"7b9166048971b1aa635d2f41cc2a141c9f60f0e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 11:24:53.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e3f864ac569ec229d8616a8755313b3bea5100bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 12:16:32.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/6/gate/openstack-tox-pep8/eb6b73e/ : SUCCESS in 3m 05s\n- openstack-tox-docs http://logs.openstack.org/27/663927/6/gate/openstack-tox-docs/c788218/html/ : SUCCESS in 2m 36s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/6/gate/monasca-tempest-python2-influxdb/c2bfd6c/ : FAILURE in 41m 50s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/6/gate/monasca-tempest-python3-influxdb/de733ea/ : SUCCESS in 37m 25s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/6/gate/monasca-tempest-python2-cassandra/a0ae1e2/ : SUCCESS in 45m 43s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/6/gate/monascalog-python2-tempest/b3e2b2d/ : SUCCESS in 42m 14s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/6/gate/monascalog-python3-tempest/09a3291/ : SUCCESS in 49m 57s","accounts_in_message":[],"_revision_number":6},{"id":"e105d8a19e9fc0d54ac156af54930d71f9e9b377","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-10 12:18:14.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"2889e445904901bc8487dd7a7e0108fe79b82eff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 13:05:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/6/check/openstack-tox-pep8/b552c61/ : SUCCESS in 2m 54s\n- openstack-tox-docs http://logs.openstack.org/27/663927/6/check/openstack-tox-docs/9dde4e9/html/ : SUCCESS in 2m 43s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/6/check/monasca-tempest-python2-influxdb/c04e357/ : SUCCESS in 36m 42s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/6/check/monasca-tempest-python3-influxdb/23549b7/ : SUCCESS in 47m 03s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/6/check/monasca-tempest-python2-cassandra/5ccc46d/ : SUCCESS in 38m 53s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/6/check/monascalog-python2-tempest/65087f9/ : SUCCESS in 39m 51s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/6/check/monascalog-python3-tempest/824b6cc/ : SUCCESS in 38m 00s\n- build-monasca-docker-image http://logs.openstack.org/27/663927/6/check/build-monasca-docker-image/1c33a98/ : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":6},{"id":"89c8c2cba2efc4643fcc588aad1f81cc147baa25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 13:06:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"73e86477558890de4ef2c4427c7ce7268e245246","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 14:02:00.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/6/gate/openstack-tox-pep8/ed09858/ : SUCCESS in 4m 05s\n- openstack-tox-docs http://logs.openstack.org/27/663927/6/gate/openstack-tox-docs/cda5731/html/ : SUCCESS in 3m 47s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/6/gate/monasca-tempest-python2-influxdb/507e421/ : FAILURE in 41m 49s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/6/gate/monasca-tempest-python3-influxdb/4a3a2f8/ : FAILURE in 46m 32s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/6/gate/monasca-tempest-python2-cassandra/2628990/ : SUCCESS in 36m 23s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/6/gate/monascalog-python2-tempest/7aae8ae/ : SUCCESS in 39m 07s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/6/gate/monascalog-python3-tempest/b137f51/ : SUCCESS in 49m 42s","accounts_in_message":[],"_revision_number":6},{"id":"42b2aa98b06fdb0bcf113e514db407e2028a5d95","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-10 14:52:20.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"bb481bf3c1f9df42ce6e41d47767110ffccb5e06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 15:48:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/6/check/openstack-tox-pep8/070a874/ : SUCCESS in 5m 08s\n- openstack-tox-docs http://logs.openstack.org/27/663927/6/check/openstack-tox-docs/3bb4f9e/html/ : SUCCESS in 2m 38s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/6/check/monasca-tempest-python2-influxdb/531f3da/ : SUCCESS in 40m 30s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/6/check/monasca-tempest-python3-influxdb/fab39b3/ : SUCCESS in 43m 06s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/6/check/monasca-tempest-python2-cassandra/a61e4fe/ : SUCCESS in 39m 55s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/6/check/monascalog-python2-tempest/ec0b264/ : SUCCESS in 48m 30s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/6/check/monascalog-python3-tempest/d5b82a4/ : SUCCESS in 42m 20s\n- build-monasca-docker-image http://logs.openstack.org/27/663927/6/check/build-monasca-docker-image/62b815c/ : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":6},{"id":"07917196069f6b8eff878cb78250a7a01cf777ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 15:48:55.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8700e936398c81a763728ce04d9c248db9804b7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:39:21.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/6/gate/openstack-tox-pep8/4a7c016/ : SUCCESS in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/27/663927/6/gate/openstack-tox-docs/aeace1c/html/ : SUCCESS in 3m 12s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/6/gate/monasca-tempest-python2-influxdb/2fa9f4a/ : FAILURE in 47m 34s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/6/gate/monasca-tempest-python3-influxdb/aa74d24/ : SUCCESS in 47m 23s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/6/gate/monasca-tempest-python2-cassandra/24bea79/ : SUCCESS in 36m 51s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/6/gate/monascalog-python2-tempest/c64fa77/ : SUCCESS in 39m 55s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/6/gate/monascalog-python3-tempest/4c70c9f/ : SUCCESS in 40m 06s","accounts_in_message":[],"_revision_number":6},{"id":"afbb4871322f3128dd28a2234cd5d6a86b3f1849","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-10 16:58:51.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"c17ed2f82455ae434d4db8466c2a4d60425e62fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:57:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/6/check/openstack-tox-pep8/715155c/ : SUCCESS in 4m 16s\n- openstack-tox-docs http://logs.openstack.org/27/663927/6/check/openstack-tox-docs/ecc434a/html/ : SUCCESS in 3m 36s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/6/check/monasca-tempest-python2-influxdb/26ab4c8/ : SUCCESS in 47m 13s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/6/check/monasca-tempest-python3-influxdb/8f0b7ed/ : SUCCESS in 41m 24s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/6/check/monasca-tempest-python2-cassandra/c3e79a2/ : SUCCESS in 38m 40s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/6/check/monascalog-python2-tempest/f9a28c7/ : SUCCESS in 50m 31s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/6/check/monascalog-python3-tempest/2a04fe7/ : SUCCESS in 42m 53s\n- build-monasca-docker-image http://logs.openstack.org/27/663927/6/check/build-monasca-docker-image/ac2d59d/ : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":6},{"id":"883b0c40bed891f45d12f2339dd6d638350774cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:57:30.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3c9c04198c2654629e8c45c695ca121d0ccbd4c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:54:34.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/27/663927/6/gate/openstack-tox-pep8/1ec8bba/ : SUCCESS in 2m 53s\n- openstack-tox-docs http://logs.openstack.org/27/663927/6/gate/openstack-tox-docs/8f02c97/html/ : SUCCESS in 3m 34s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/27/663927/6/gate/monasca-tempest-python2-influxdb/2447ff7/ : SUCCESS in 44m 50s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/27/663927/6/gate/monasca-tempest-python3-influxdb/741ee2d/ : SUCCESS in 34m 28s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/27/663927/6/gate/monasca-tempest-python2-cassandra/45e0df7/ : SUCCESS in 35m 01s\n- monascalog-python2-tempest http://logs.openstack.org/27/663927/6/gate/monascalog-python2-tempest/c32ba39/ : SUCCESS in 40m 35s\n- monascalog-python3-tempest http://logs.openstack.org/27/663927/6/gate/monascalog-python3-tempest/ae431f0/ : SUCCESS in 48m 37s","accounts_in_message":[],"_revision_number":6},{"id":"88a0072ae14b46c1b7ce2d8c3b1a41380543c2e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:54:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ffd53171589cff0575666e072cd1b62c323c7ce7","revisions":{"bd3ddfd145ded9df882ad6ebed6932deefb2f97a":{"kind":"REWORK","_number":1,"created":"2019-06-07 12:51:39.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/27/663927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-tempest-plugin","ref":"refs/changes/27/663927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/1"}}},"commit":{"parents":[{"commit":"46a97d9c523613eff73b6aacc0e5dc2fb57452e9","subject":"Update alarms count tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/46a97d9c523613eff73b6aacc0e5dc2fb57452e9"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-06-07 12:07:24.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-06-07 12:44:34.000000000","tz":120},"subject":"Encode metric names and dimensions in test_metrics","message":"Encode metric names and dimensions in test_metrics\n\nThe change explicitly encodes all metric names and dimensions. It makes\nthe tests more robust and avoids failures when these contain non-ASCII\ncharacters.\n\nThe code uses oslo_utils.encodeutils to correctly handle Python 2 and\nPython 3 deployments.\n\nChange-Id: I87c38faf41f7c1bff72d45d238a6f38751fad483\nStory: 2005841\nTask: 33613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/bd3ddfd145ded9df882ad6ebed6932deefb2f97a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/bd3ddfd145ded9df882ad6ebed6932deefb2f97a"}]},"branch":"refs/heads/master"},"73c71649d13d75afba9d5991e5bf729059c549d9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-24 08:51:51.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/27/663927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-tempest-plugin","ref":"refs/changes/27/663927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/2"}}},"commit":{"parents":[{"commit":"f90c73911d16556f07a17f3776bccac325e3b3f5","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/f90c73911d16556f07a17f3776bccac325e3b3f5"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-06-07 12:07:24.000000000","tz":120},"committer":{"name":"Witold Bedyk","email":"witold.bedyk@suse.com","date":"2019-06-24 08:51:51.000000000","tz":0},"subject":"Encode metric names and dimensions in test_metrics","message":"Encode metric names and dimensions in test_metrics\n\nThe change explicitly encodes all metric names and dimensions. It makes\nthe tests more robust and avoids failures when these contain non-ASCII\ncharacters.\n\nThe code uses oslo_utils.encodeutils to correctly handle Python 2 and\nPython 3 deployments.\n\nChange-Id: I87c38faf41f7c1bff72d45d238a6f38751fad483\nStory: 2005841\nTask: 33613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/73c71649d13d75afba9d5991e5bf729059c549d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/73c71649d13d75afba9d5991e5bf729059c549d9"}]},"branch":"refs/heads/master"},"50b99047d85d1b1353eb5b8fe8aaa1f326404501":{"kind":"REWORK","_number":3,"created":"2019-06-26 12:08:26.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/27/663927/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-tempest-plugin","ref":"refs/changes/27/663927/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/3"}}},"commit":{"parents":[{"commit":"46a97d9c523613eff73b6aacc0e5dc2fb57452e9","subject":"Update alarms count tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/46a97d9c523613eff73b6aacc0e5dc2fb57452e9"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-06-07 12:07:24.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-06-26 12:08:16.000000000","tz":120},"subject":"Encode metric names and dimensions in test_metrics","message":"Encode metric names and dimensions in test_metrics\n\nThe change explicitly encodes all metric names and dimensions. It makes\nthe tests more robust and avoids failures when these contain non-ASCII\ncharacters.\n\nThe code uses oslo_utils.to_utf() to correctly handle Python 2 and\nPython 3 deployments.\n\nChange-Id: I87c38faf41f7c1bff72d45d238a6f38751fad483\nStory: 2005841\nTask: 33613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/50b99047d85d1b1353eb5b8fe8aaa1f326404501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/50b99047d85d1b1353eb5b8fe8aaa1f326404501"}]},"branch":"refs/heads/master"},"30b4b41c7344e9caae528a707342387c437f3404":{"kind":"REWORK","_number":4,"created":"2019-07-02 12:19:50.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/27/663927/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-tempest-plugin","ref":"refs/changes/27/663927/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/4"}}},"commit":{"parents":[{"commit":"46a97d9c523613eff73b6aacc0e5dc2fb57452e9","subject":"Update alarms count tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/46a97d9c523613eff73b6aacc0e5dc2fb57452e9"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-06-07 12:07:24.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2019-07-02 12:17:20.000000000","tz":120},"subject":"Decode metric names and dimensions in test_metrics","message":"Decode metric names and dimensions in test_metrics\n\nThe change explicitly decodes all metric names and dimensions. It makes\nthe tests more robust and avoids failures when these contain non-ASCII\ncharacters.\n\nThe code uses oslo_utils.encodeutils to correctly handle Python 2 and\nPython 3 deployments.\n\nChange-Id: I87c38faf41f7c1bff72d45d238a6f38751fad483\nStory: 2005841\nTask: 33613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/30b4b41c7344e9caae528a707342387c437f3404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/30b4b41c7344e9caae528a707342387c437f3404"}]},"branch":"refs/heads/master"},"357dc8a6b21d8a98eedbefdf0d36c57756370932":{"kind":"REWORK","_number":5,"created":"2019-07-03 08:43:52.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/27/663927/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-tempest-plugin","ref":"refs/changes/27/663927/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/5"}}},"commit":{"parents":[{"commit":"46a97d9c523613eff73b6aacc0e5dc2fb57452e9","subject":"Update alarms count tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/46a97d9c523613eff73b6aacc0e5dc2fb57452e9"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-06-07 12:07:24.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-07-03 08:33:32.000000000","tz":120},"subject":"Use unicode literals in test_metrics","message":"Use unicode literals in test_metrics\n\nThe change unifies the usage of text string literals in the package and\ndeclares them as unicode. It simplifies comparing test values with\nactual results across Python 2 and 3 deployments. It also fixes the bug\ndescribed in the user story.\n\nChange-Id: I87c38faf41f7c1bff72d45d238a6f38751fad483\nStory: 2005841\nTask: 33613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/357dc8a6b21d8a98eedbefdf0d36c57756370932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/357dc8a6b21d8a98eedbefdf0d36c57756370932"}]},"branch":"refs/heads/master"},"ffd53171589cff0575666e072cd1b62c323c7ce7":{"kind":"REWORK","_number":6,"created":"2019-07-03 08:57:22.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/27/663927/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-tempest-plugin","ref":"refs/changes/27/663927/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-tempest-plugin refs/changes/27/663927/6"}}},"commit":{"parents":[{"commit":"46a97d9c523613eff73b6aacc0e5dc2fb57452e9","subject":"Update alarms count tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/46a97d9c523613eff73b6aacc0e5dc2fb57452e9"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-06-07 12:07:24.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-07-03 08:57:13.000000000","tz":120},"subject":"Use unicode literals in test_metrics","message":"Use unicode literals in test_metrics\n\nThe change unifies the usage of text string literals in the package and\ndeclares them as unicode. It simplifies comparing test values with\nactual results across Python 2 and 3 deployments. It also fixes the bug\ndescribed in the user story.\n\nChange-Id: I87c38faf41f7c1bff72d45d238a6f38751fad483\nStory: 2005841\nTask: 33613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/ffd53171589cff0575666e072cd1b62c323c7ce7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-tempest-plugin/commit/ffd53171589cff0575666e072cd1b62c323c7ce7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
