)]}'
{"id":"openstack%2Fmonasca-thresh~649298","triplet_id":"openstack%2Fmonasca-thresh~master~If3323a18ac3697a92ca3f06b0db154517582d7aa","project":"openstack/monasca-thresh","branch":"master","topic":"feature/piotrowskim/docker","hashtags":[],"change_id":"If3323a18ac3697a92ca3f06b0db154517582d7aa","subject":"Create Docker image and build in Zuul for Monasca-Thresh","status":"MERGED","created":"2019-04-02 12:26:56.000000000","updated":"2019-09-25 09:36:52.000000000","submitted":"2019-09-25 09:36:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":33,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"649298-1569404212525-e03b07bf","meta_rev_id":"9e81f10771f8d870f7e2aab187268947a8857f18","_number":649298,"virtual_id_number":649298,"owner":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 09:36:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":2,"date":"2019-09-19 11:27:35.000000000","_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-16 09:03:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":2,"date":"2019-09-24 16:04:15.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":1,"date":"2019-09-25 09:25:45.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-09 09:36:58.000000000","updated_by":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"reviewer":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"state":"REVIEWER"},{"updated":"2019-07-17 07:34:55.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2019-09-16 09:03:15.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2019-09-19 11:27:35.000000000","updated_by":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"reviewer":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"state":"REVIEWER"},{"updated":"2019-09-25 09:25:45.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-09-25 09:36:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d3e0b36ff8437559079d4b57ee6691c2c14fad6","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-04-02 12:26:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ace5f29663b85c7c56bf529217a5ed9e17b0ceba","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-04-02 12:53:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4f9f0ea8dd3df39b16c6c1967ec124ee1d968a73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 13:49:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/2/check/monasca-tempest-python3-influxdb/8d688ab/ : SUCCESS in 34m 31s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/2/check/monasca-tempest-python3-cassandra/21feb19/ : FAILURE in 28m 09s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/2/check/monasca-tempest-java-influxdb/6203af5/ : FAILURE in 50m 39s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/2/check/legacy-monasca-thresh-maven-build/1ad3573/ : SUCCESS in 6m 30s\n- docker-build-monasca-service http://logs.openstack.org/98/649298/2/check/docker-build-monasca-service/8a461aa/ : FAILURE in 2m 17s","accounts_in_message":[],"_revision_number":2},{"id":"c539034508a170d408a79c02554b7fb904a607fc","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-04-03 09:09:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b35a80d560b711de64e72d4cb9593fb98f8e7fd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-03 10:16:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/3/check/monasca-tempest-python3-influxdb/10a6e63/ : SUCCESS in 34m 13s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/3/check/monasca-tempest-python3-cassandra/4afc079/ : FAILURE in 20m 13s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/3/check/monasca-tempest-java-influxdb/d4d90c3/ : FAILURE in 58m 56s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/3/check/legacy-monasca-thresh-maven-build/8e21e55/ : SUCCESS in 5m 44s\n- docker-build-monasca-service http://logs.openstack.org/98/649298/3/check/docker-build-monasca-service/571c283/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":3},{"id":"2f8d60666ccaf678a1e961526d0adda3fb82a23a","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-04-04 10:03:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5769c3acee8cca90f30717a4e14de1f225395297","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 10:57:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/4/check/monasca-tempest-python3-influxdb/e0e397a/ : SUCCESS in 46m 34s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/4/check/monasca-tempest-python3-cassandra/6c0765e/ : FAILURE in 25m 20s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/4/check/monasca-tempest-java-influxdb/a795551/ : FAILURE in 44m 04s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/4/check/legacy-monasca-thresh-maven-build/4232f4e/ : SUCCESS in 5m 42s\n- docker-build-monasca-service http://logs.openstack.org/98/649298/4/check/docker-build-monasca-service/c9ef782/ : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":4},{"id":"f44940cef2ad797fc3a80ef4a4a085d3a70ed105","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-04-09 13:15:19.000000000","message":"Patch Set 4: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4e8544041c89614700e601d1fac7ec41a4a8e4f6","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-04-10 08:28:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d2d3d0379f95d694e69bca631627d731bef2a084","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 09:21:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/5/check/monasca-tempest-python3-influxdb/a0583d5/ : SUCCESS in 37m 49s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/5/check/monasca-tempest-python3-cassandra/efff51f/ : FAILURE in 27m 01s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/5/check/monasca-tempest-java-influxdb/472ad6f/ : FAILURE in 47m 58s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/5/check/legacy-monasca-thresh-maven-build/622b6c8/ : SUCCESS in 5m 16s\n- docker-build-monasca-service http://logs.openstack.org/98/649298/5/check/docker-build-monasca-service/0f43af4/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":5},{"id":"804ff6c0b7b3d49063bb81080bd1f42807e9af4c","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-04-23 14:05:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2abd95bfcb9d53ba476291773cd88b8d1bd5f827","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 15:38:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/6/check/monasca-tempest-python3-influxdb/0a6c22f/ : SUCCESS in 37m 13s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/6/check/monasca-tempest-python3-cassandra/6fc4119/ : FAILURE in 23m 20s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/6/check/monasca-tempest-java-influxdb/4625e79/ : FAILURE in 47m 27s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/6/check/legacy-monasca-thresh-maven-build/bb945a5/ : SUCCESS in 9m 12s","accounts_in_message":[],"_revision_number":6},{"id":"99b2505303f79931d71a07b39c13fea4dad2230e","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-05-10 06:25:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"38160e0181ba87d4476b045c47567399ec31d160","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 07:53:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/7/check/monasca-tempest-python3-influxdb/1a0593f/ : FAILURE in 1h 22m 57s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/7/check/monasca-tempest-python3-cassandra/08e7e25/ : FAILURE in 27m 22s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/7/check/monasca-tempest-java-influxdb/b51ec17/ : FAILURE in 1h 22m 59s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/7/check/legacy-monasca-thresh-maven-build/5ecf010/ : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":7},{"id":"b5fee831958caa3e30c716051bf6c5bec0fddd5d","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-05-15 09:49:02.000000000","message":"Patch Set 7: Code-Review-1\n\n(17 comments)","accounts_in_message":[],"_revision_number":7},{"id":"234865760b355bd1cf1b5f78dbb5f154ea9a5510","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-05-16 06:51:54.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cec970aea210d386706fbb466c86f7f2cc5dbae0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 07:45:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/8/check/monasca-tempest-python3-influxdb/603f171/ : SUCCESS in 39m 17s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/8/check/monasca-tempest-python3-cassandra/8d53fdd/ : FAILURE in 24m 20s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/8/check/monasca-tempest-java-influxdb/4b4211a/ : FAILURE in 47m 45s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/8/check/legacy-monasca-thresh-maven-build/c5fd538/ : SUCCESS in 5m 04s\n- build-monasca-docker-image http://logs.openstack.org/98/649298/8/check/build-monasca-docker-image/f29266d/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":8},{"id":"c41a9b879423f4619b8936eb2dc33b0e7fcba70d","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-05-28 12:47:32.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4dad6c36a4ab2d0f9ad53471104c9ed90fc08cd6","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-05-28 13:23:34.000000000","message":"Patch Set 9: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a2939dc931a2e3f875a34c7a622f8d1d063370a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 13:39:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/9/check/monasca-tempest-python3-influxdb/b61dfa5/ : SUCCESS in 41m 10s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/9/check/monasca-tempest-python3-cassandra/127afd0/ : FAILURE in 23m 15s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/9/check/monasca-tempest-java-influxdb/1851715/ : FAILURE in 43m 18s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/9/check/legacy-monasca-thresh-maven-build/ce01163/ : SUCCESS in 4m 41s\n- build-monasca-docker-image http://logs.openstack.org/98/649298/9/check/build-monasca-docker-image/9e41ea2/ : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":9},{"id":"2318fa298ef9868431dbe1c0a084bb6773645992","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-05-29 07:50:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8173d45ef606d3435cc1362dce5e124080f41b7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 08:43:32.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/10/check/monasca-tempest-python3-influxdb/3e8dca4/ : SUCCESS in 42m 16s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/10/check/monasca-tempest-python3-cassandra/09d147c/ : FAILURE in 26m 18s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/10/check/monasca-tempest-java-influxdb/4be7c3c/ : FAILURE in 47m 12s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/10/check/legacy-monasca-thresh-maven-build/4d19799/ : SUCCESS in 5m 27s\n- build-monasca-docker-image http://logs.openstack.org/98/649298/10/check/build-monasca-docker-image/2bc6bab/ : FAILURE in 1m 52s","accounts_in_message":[],"_revision_number":10},{"id":"528487edfe5bcc3839fac1d46116c1c7014057c3","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-05-29 12:28:56.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0c0d5a64810a6b6ae2cc1e588fcccc4678831619","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-05-29 12:29:36.000000000","message":"Patch Set 11: Code-Review-1\n\nIt still needs readme","accounts_in_message":[],"_revision_number":11},{"id":"dff7e756a5d7adfa7127d436d4fb93c42e0c1482","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-05-29 12:55:05.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"9827fb114784cb3714a251d75c39704e12882262","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 13:51:30.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/12/check/monasca-tempest-python3-influxdb/5821ef4/ : SUCCESS in 36m 41s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/12/check/monasca-tempest-python3-cassandra/aa38af5/ : FAILURE in 26m 49s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/12/check/monasca-tempest-java-influxdb/6610fdd/ : FAILURE in 50m 30s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/12/check/legacy-monasca-thresh-maven-build/455e52d/ : SUCCESS in 5m 14s\n- build-monasca-docker-image http://logs.openstack.org/98/649298/12/check/build-monasca-docker-image/d1a2782/ : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":12},{"id":"94f06e4f03ba276811c836e75c573f977076b9be","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-05-30 11:28:04.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e678e4683563d2a80826609220b2e040458fef30","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-06-05 10:37:42.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"ebeef2f4119de2b0b1e1d7eab9951674a667540a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 11:30:31.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/13/check/monasca-tempest-python3-influxdb/15c5ce1/ : SUCCESS in 42m 20s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/13/check/monasca-tempest-python3-cassandra/62c3ad1/ : FAILURE in 27m 13s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/13/check/monasca-tempest-java-influxdb/9370194/ : FAILURE in 49m 25s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/13/check/legacy-monasca-thresh-maven-build/1d25167/ : SUCCESS in 5m 15s\n- build-monasca-docker-image http://logs.openstack.org/98/649298/13/check/build-monasca-docker-image/7467f76/ : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":13},{"id":"255d013a81359c7f8d1913bfa838d5374f89aad0","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-06-17 12:57:00.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"56a73c7a4314ecb9653937e0206a269282172b80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 13:51:18.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/14/check/monasca-tempest-python3-influxdb/9f36560/ : FAILURE in 40m 22s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/14/check/monasca-tempest-python3-cassandra/0a504e0/ : FAILURE in 22m 09s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/14/check/monasca-tempest-java-influxdb/f577734/ : FAILURE in 45m 44s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/14/check/legacy-monasca-thresh-maven-build/2d504e9/ : SUCCESS in 5m 25s\n- build-monasca-docker-image http://logs.openstack.org/98/649298/14/check/build-monasca-docker-image/d59e370/ : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":14},{"id":"b06f82c65f43a50256f160b1a0ed9c3d259ca37c","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2019-06-18 06:46:57.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"92ae3a179a67661f871a68bb4c0ddb4ce8659f0e","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2019-06-18 06:57:30.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"30cbba3bf8b34e040a2ae9ecdf09780db828f371","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-06-18 13:13:37.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"b58f0f6623b16b69034a0e35893d4043b902a6e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 14:08:26.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/15/check/monasca-tempest-python3-influxdb/65c6b5e/ : FAILURE in 39m 58s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/15/check/monasca-tempest-python3-cassandra/19c0e3d/ : FAILURE in 24m 58s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/15/check/monasca-tempest-java-influxdb/673d0b3/ : FAILURE in 49m 04s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/15/check/legacy-monasca-thresh-maven-build/da31d77/ : SUCCESS in 7m 16s\n- build-monasca-docker-image http://logs.openstack.org/98/649298/15/check/build-monasca-docker-image/b5f863a/ : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":15},{"id":"8e5bca6db4fc356bec7bb0fc78c2ecf1c264975c","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-07-01 12:38:48.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"ca5fafc803ccbb564bb99da7fcf83744d5dda930","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-07-01 12:39:46.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"3ecacbaced97b16ca5753109325e3127e1ff2b0e","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-07-01 12:39:58.000000000","message":"Patch Set 16: Workflow-1","accounts_in_message":[],"_revision_number":16},{"id":"0c68d7c33b78f6b1edc369fa7709d776e0c461b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 13:29:56.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/16/check/monasca-tempest-python3-influxdb/344d291/ : FAILURE in 49m 16s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/16/check/monasca-tempest-python3-cassandra/c06472f/ : FAILURE in 21m 09s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/16/check/monasca-tempest-java-influxdb/5fe54be/ : FAILURE in 45m 02s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/16/check/legacy-monasca-thresh-maven-build/fda4887/ : SUCCESS in 5m 28s\n- build-monasca-docker-image http://logs.openstack.org/98/649298/16/check/build-monasca-docker-image/e00ddeb/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":16},{"id":"1c4ede994d6434245e2e00837e91895cd200d6e8","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-07-09 09:36:24.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"8fb09646aa44a7584ff0f276c4ee1f7e1e68ec3e","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-07-09 09:36:58.000000000","message":"Patch Set 17: Workflow-1","accounts_in_message":[],"_revision_number":17},{"id":"4598f44bc1fbb61613ecbaaae8960d467e431c69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 10:36:25.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/17/check/monasca-tempest-python3-influxdb/09c205e/ : SUCCESS in 39m 55s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/17/check/monasca-tempest-python3-cassandra/4ed6b04/ : FAILURE in 25m 30s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/17/check/monasca-tempest-java-influxdb/99a3b3d/ : FAILURE in 54m 12s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/17/check/legacy-monasca-thresh-maven-build/0b343d9/ : SUCCESS in 5m 26s\n- build-monasca-docker-image http://logs.openstack.org/98/649298/17/check/build-monasca-docker-image/6836b6f/ : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":17},{"id":"f07aaedb4a7f83377976d4b32ff1c3d6c2ddad04","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-07-16 14:29:25.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"9c9752f97868df8305b4a3fd1bc5c5ed0ab700ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:27:46.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/18/check/monasca-tempest-python3-influxdb/ed8c445/ : SUCCESS in 38m 46s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/18/check/monasca-tempest-python3-cassandra/8952d7a/ : FAILURE in 24m 13s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/18/check/monasca-tempest-java-influxdb/46353cb/ : FAILURE in 53m 16s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/18/check/legacy-monasca-thresh-maven-build/3668620/ : SUCCESS in 6m 43s\n- build-monasca-docker-image http://logs.openstack.org/98/649298/18/check/build-monasca-docker-image/4178915/ : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":18},{"id":"05052afe987c421eb0dabc2bb5632f138d5991d9","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2019-07-17 07:34:55.000000000","message":"Patch Set 18: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"4f42f91173754d91dc65deb0e96d38e494e7d5d9","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-07-25 10:19:47.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"d0cacb4cb597b0debe4e44d98cabb6345f9484c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 11:20:33.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/19/check/monasca-tempest-python3-influxdb/7467a17/ : SUCCESS in 40m 40s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/19/check/monasca-tempest-python3-cassandra/4ceb809/ : FAILURE in 24m 09s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/19/check/monasca-tempest-java-influxdb/827a6fc/ : FAILURE in 57m 07s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/19/check/legacy-monasca-thresh-maven-build/d4fa254/ : SUCCESS in 5m 01s\n- build-monasca-docker-image http://logs.openstack.org/98/649298/19/check/build-monasca-docker-image/58f44f6/ : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":19},{"id":"7edfff54bba502af4135a7f40ad185eb6936a1f7","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-07-25 14:10:34.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"f8ac50f7381bdd60f650cfaea0a122c512242107","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 15:09:03.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb http://logs.openstack.org/98/649298/20/check/monasca-tempest-python3-influxdb/73a6961/ : SUCCESS in 41m 59s (non-voting)\n- monasca-tempest-python3-cassandra http://logs.openstack.org/98/649298/20/check/monasca-tempest-python3-cassandra/c3ceb88/ : FAILURE in 34m 58s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/98/649298/20/check/monasca-tempest-java-influxdb/daba1f3/ : FAILURE in 54m 18s (non-voting)\n- legacy-monasca-thresh-maven-build http://logs.openstack.org/98/649298/20/check/legacy-monasca-thresh-maven-build/e0ccd52/ : SUCCESS in 7m 16s\n- build-monasca-docker-image http://logs.openstack.org/98/649298/20/check/build-monasca-docker-image/00dd4ef/ : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":20},{"id":"723ca0895cf49eb3f372dbb673ae44312e195f0a","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-08-05 14:00:17.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"c0df31c879f7d7f9b72176991604f9ec00839c83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 15:06:37.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb https://logs.opendev.org/98/649298/21/check/monasca-tempest-python3-influxdb/b77b89c/ : SUCCESS in 40m 01s (non-voting)\n- monasca-tempest-python3-cassandra https://logs.opendev.org/98/649298/21/check/monasca-tempest-python3-cassandra/b19e6ea/ : FAILURE in 22m 24s (non-voting)\n- monasca-tempest-java-influxdb https://logs.opendev.org/98/649298/21/check/monasca-tempest-java-influxdb/ea38cdc/ : FAILURE in 48m 48s (non-voting)\n- legacy-monasca-thresh-maven-build https://logs.opendev.org/98/649298/21/check/legacy-monasca-thresh-maven-build/afb7526/ : SUCCESS in 6m 38s\n- build-monasca-docker-image https://logs.opendev.org/98/649298/21/check/build-monasca-docker-image/e4da775/ : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":21},{"id":"698a056aaf7c5d9fc776114edc04f45500cba813","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-08-06 13:07:24.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"246b3a9703d85d776881bfc9fc60594199f80a22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 14:16:04.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb https://logs.opendev.org/98/649298/22/check/monasca-tempest-python3-influxdb/9b0be4b/ : SUCCESS in 44m 00s (non-voting)\n- monasca-tempest-python3-cassandra https://logs.opendev.org/98/649298/22/check/monasca-tempest-python3-cassandra/2ee9965/ : FAILURE in 26m 10s (non-voting)\n- monasca-tempest-java-influxdb https://logs.opendev.org/98/649298/22/check/monasca-tempest-java-influxdb/ae8546a/ : FAILURE in 1h 01m 35s (non-voting)\n- legacy-monasca-thresh-maven-build https://logs.opendev.org/98/649298/22/check/legacy-monasca-thresh-maven-build/d9d245d/ : SUCCESS in 6m 39s\n- build-monasca-docker-image https://logs.opendev.org/98/649298/22/check/build-monasca-docker-image/f533a64/ : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":22},{"id":"fa636bb44f8aa6810cf7ad4c955247c4f11ef99e","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-08-19 09:09:18.000000000","message":"Patch Set 22: Code-Review+1","accounts_in_message":[],"_revision_number":22},{"id":"5937c1e604d61eb7e32f2dceceb0ffeb6ee17054","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-08-29 08:09:26.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"1472f4b3ecb2b6d35da8558520c5351d68329070","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 09:05:23.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/a0747e60f3c64b118a7eeb0871091ee8 : SUCCESS in 41m 49s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/74383d2cecfd47c3b4a81cd098f67219 : FAILURE in 27m 13s (non-voting)\n- monasca-tempest-java-influxdb https://zuul.opendev.org/t/openstack/build/43333aef597b42cea4937784e3ed9845 : FAILURE in 50m 03s (non-voting)\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/5a2c6ebdce6e45cfbf31cf62f016da68 : SUCCESS in 6m 04s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/572d82bccd644d5198f9448392471822 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":23},{"id":"ca4343b6eeb067bc470f1cc4eabf82ddc5305e96","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-09-16 09:03:15.000000000","message":"Patch Set 23: Code-Review+1","accounts_in_message":[],"_revision_number":23},{"id":"0643105ffbb9cc87f98bb9e1edf99d48cf5ec251","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-09-18 15:18:45.000000000","message":"Patch Set 23: Review-Priority+1","accounts_in_message":[],"_revision_number":23},{"id":"f502494393db97bb8757c2e526694f5866a6f9e8","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2019-09-19 11:27:35.000000000","message":"Patch Set 23: Code-Review+2","accounts_in_message":[],"_revision_number":23},{"id":"21961137bf84947399fe6122296641a580afceef","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-09-24 16:04:15.000000000","message":"Patch Set 23: Code-Review+2\n\nSuccessfully built the image and run tempest tests with the new container. All tests passed.","accounts_in_message":[],"_revision_number":23},{"id":"0ccbdff049ed561d55af6f9125eb4339d13b6251","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-09-25 09:25:45.000000000","message":"Patch Set 23: Workflow+1","accounts_in_message":[],"_revision_number":23},{"id":"0d7f6454abb3169706b01f81f449abae2fce9cef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 09:25:59.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"cb7c73bde22644e16252ecc3ed1095972bbbeb69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 09:36:52.000000000","message":"Patch Set 23: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/838fca19cacf4f99b5e0760ac0189af2 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":23},{"id":"f8ac7e1d57a0bac86673755ba8a4c29cac31e065","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 09:36:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":23}],"current_revision_number":23,"current_revision":"5cb9f94387dcbd76a7d261b047b65aa63aeb5092","revisions":{"02a3ce0aedda80c47ddfe73733488cb316930039":{"kind":"REWORK","_number":1,"created":"2019-04-02 12:26:56.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/98/649298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/1"}}},"commit":{"parents":[{"commit":"8c976339a79d8be6396503821dc4d8f384633a0b","subject":"Use storm version 1.1.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/8c976339a79d8be6396503821dc4d8f384633a0b"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:26:42.000000000","tz":120},"subject":"[WiP] Create Docker image and build in Zuul","message":"[WiP] Create Docker image and build in Zuul\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\nStory: 2001694\nTask: 30289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/02a3ce0aedda80c47ddfe73733488cb316930039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/02a3ce0aedda80c47ddfe73733488cb316930039"}]},"branch":"refs/heads/master"},"4c62255b16f6dbaf8b5c97fbd1d50e8f17741e05":{"kind":"REWORK","_number":2,"created":"2019-04-02 12:53:05.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/98/649298/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/2"}}},"commit":{"parents":[{"commit":"8c976339a79d8be6396503821dc4d8f384633a0b","subject":"Use storm version 1.1.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/8c976339a79d8be6396503821dc4d8f384633a0b"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:52:54.000000000","tz":120},"subject":"[WiP] Create Docker image and build in Zuul","message":"[WiP] Create Docker image and build in Zuul\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\nStory: 2001694\nTask: 30289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/4c62255b16f6dbaf8b5c97fbd1d50e8f17741e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/4c62255b16f6dbaf8b5c97fbd1d50e8f17741e05"}]},"branch":"refs/heads/master"},"14dac95715d1788edfcd31031569095f15d91936":{"kind":"REWORK","_number":3,"created":"2019-04-03 09:09:45.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/98/649298/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/3"}}},"commit":{"parents":[{"commit":"8c976339a79d8be6396503821dc4d8f384633a0b","subject":"Use storm version 1.1.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/8c976339a79d8be6396503821dc4d8f384633a0b"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-03 09:09:14.000000000","tz":120},"subject":"[WiP] Create Docker image and build in Zuul","message":"[WiP] Create Docker image and build in Zuul\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\nStory: 2001694\nTask: 30289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/14dac95715d1788edfcd31031569095f15d91936"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/14dac95715d1788edfcd31031569095f15d91936"}]},"branch":"refs/heads/master"},"36a047cb8b128ac2277c6bec43e9d7e0abb2e13e":{"kind":"REWORK","_number":4,"created":"2019-04-04 10:03:04.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/98/649298/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/4"}}},"commit":{"parents":[{"commit":"8c976339a79d8be6396503821dc4d8f384633a0b","subject":"Use storm version 1.1.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/8c976339a79d8be6396503821dc4d8f384633a0b"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-04 10:02:44.000000000","tz":120},"subject":"Create Docker image and build in Zuul","message":"Create Docker image and build in Zuul\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\nStory: 2001694\nTask: 30289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/36a047cb8b128ac2277c6bec43e9d7e0abb2e13e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/36a047cb8b128ac2277c6bec43e9d7e0abb2e13e"}]},"branch":"refs/heads/master"},"db9e0cacddd960ad499699c49c88e5e6642061ee":{"kind":"REWORK","_number":5,"created":"2019-04-10 08:28:24.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/98/649298/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/5"}}},"commit":{"parents":[{"commit":"8c976339a79d8be6396503821dc4d8f384633a0b","subject":"Use storm version 1.1.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/8c976339a79d8be6396503821dc4d8f384633a0b"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-10 08:27:58.000000000","tz":120},"subject":"Create Docker image and build in Zuul","message":"Create Docker image and build in Zuul\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\nStory: 2001694\nTask: 30289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/db9e0cacddd960ad499699c49c88e5e6642061ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/db9e0cacddd960ad499699c49c88e5e6642061ee"}]},"branch":"refs/heads/master"},"6545a95b410383d7289662caa64e4dbcbd5ac15e":{"kind":"REWORK","_number":6,"created":"2019-04-23 14:05:03.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/98/649298/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/6"}}},"commit":{"parents":[{"commit":"8c976339a79d8be6396503821dc4d8f384633a0b","subject":"Use storm version 1.1.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/8c976339a79d8be6396503821dc4d8f384633a0b"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-23 14:04:51.000000000","tz":120},"subject":"Create Docker image and build in Zuul","message":"Create Docker image and build in Zuul\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\nStory: 2001694\nTask: 30289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/6545a95b410383d7289662caa64e4dbcbd5ac15e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/6545a95b410383d7289662caa64e4dbcbd5ac15e"}]},"branch":"refs/heads/master"},"9669d672877718c9a2370e426315de72c892e7fc":{"kind":"REWORK","_number":7,"created":"2019-05-10 06:25:24.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/98/649298/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/7"}}},"commit":{"parents":[{"commit":"8c976339a79d8be6396503821dc4d8f384633a0b","subject":"Use storm version 1.1.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/8c976339a79d8be6396503821dc4d8f384633a0b"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-05-10 06:24:49.000000000","tz":120},"subject":"Create Docker image and build in Zuul","message":"Create Docker image and build in Zuul\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\nStory: 2001694\nTask: 30289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/9669d672877718c9a2370e426315de72c892e7fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/9669d672877718c9a2370e426315de72c892e7fc"}]},"branch":"refs/heads/master"},"55913cd1a8291675f484f553ac889bd1da11288d":{"kind":"REWORK","_number":8,"created":"2019-05-16 06:51:54.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/98/649298/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/8"}}},"commit":{"parents":[{"commit":"de7f4c719c1e4db08c6dbc1aa75779c3c9aeb577","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/de7f4c719c1e4db08c6dbc1aa75779c3c9aeb577"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-05-16 06:51:28.000000000","tz":120},"subject":"Create Docker image and build in Zuul","message":"Create Docker image and build in Zuul\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\nStory: 2001694\nTask: 30289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/55913cd1a8291675f484f553ac889bd1da11288d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/55913cd1a8291675f484f553ac889bd1da11288d"}]},"branch":"refs/heads/master"},"55a6679cd83e547a321186d06dd6563b2a6d1108":{"kind":"REWORK","_number":9,"created":"2019-05-28 12:47:32.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/98/649298/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/9"}}},"commit":{"parents":[{"commit":"de7f4c719c1e4db08c6dbc1aa75779c3c9aeb577","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/de7f4c719c1e4db08c6dbc1aa75779c3c9aeb577"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-05-28 12:47:06.000000000","tz":120},"subject":"[WiP] Create Docker image and build in Zuul","message":"[WiP] Create Docker image and build in Zuul\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\nStory: 2001694\nTask: 30289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/55a6679cd83e547a321186d06dd6563b2a6d1108"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/55a6679cd83e547a321186d06dd6563b2a6d1108"}]},"branch":"refs/heads/master"},"2f134948918443e19f14f371c32249fc3b4afaaa":{"kind":"REWORK","_number":10,"created":"2019-05-29 07:50:40.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/98/649298/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/10"}}},"commit":{"parents":[{"commit":"de7f4c719c1e4db08c6dbc1aa75779c3c9aeb577","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/de7f4c719c1e4db08c6dbc1aa75779c3c9aeb577"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Michal Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-05-29 07:43:09.000000000","tz":120},"subject":"Create Docker image and build in Zuul","message":"Create Docker image and build in Zuul\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\nStory: 2001694\nTask: 30289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/2f134948918443e19f14f371c32249fc3b4afaaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/2f134948918443e19f14f371c32249fc3b4afaaa"}]},"branch":"refs/heads/master"},"887f48671961319d674f750a32fbe90c989529fc":{"kind":"REWORK","_number":11,"created":"2019-05-29 12:28:56.000000000","uploader":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"ref":"refs/changes/98/649298/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/11"}}},"commit":{"parents":[{"commit":"de7f4c719c1e4db08c6dbc1aa75779c3c9aeb577","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/de7f4c719c1e4db08c6dbc1aa75779c3c9aeb577"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","date":"2019-05-29 12:28:38.000000000","tz":120},"subject":"Create Docker image and build in Zuul","message":"Create Docker image and build in Zuul\n\nStory: 2001694\nTask: 30289\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/887f48671961319d674f750a32fbe90c989529fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/887f48671961319d674f750a32fbe90c989529fc"}]},"branch":"refs/heads/master"},"9e389bf9690753df0a9c7fab6806b4a67e154581":{"kind":"REWORK","_number":12,"created":"2019-05-29 12:55:05.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/98/649298/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/12"}}},"commit":{"parents":[{"commit":"de7f4c719c1e4db08c6dbc1aa75779c3c9aeb577","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/de7f4c719c1e4db08c6dbc1aa75779c3c9aeb577"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-05-29 12:54:52.000000000","tz":120},"subject":"Create Docker image and build in Zuul","message":"Create Docker image and build in Zuul\n\nStory: 2001694\nTask: 30289\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/9e389bf9690753df0a9c7fab6806b4a67e154581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/9e389bf9690753df0a9c7fab6806b4a67e154581"}]},"branch":"refs/heads/master"},"72c4f22a2548981925fc49cfb9aafa98865f4f66":{"kind":"REWORK","_number":13,"created":"2019-06-05 10:37:42.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/98/649298/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/13"}}},"commit":{"parents":[{"commit":"de7f4c719c1e4db08c6dbc1aa75779c3c9aeb577","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/de7f4c719c1e4db08c6dbc1aa75779c3c9aeb577"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-06-05 10:36:55.000000000","tz":120},"subject":"Create Docker image and build in Zuul for Monasca-Thresh","message":"Create Docker image and build in Zuul for Monasca-Thresh\n\nStory: 2001694\nTask: 30289\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/72c4f22a2548981925fc49cfb9aafa98865f4f66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/72c4f22a2548981925fc49cfb9aafa98865f4f66"}]},"branch":"refs/heads/master"},"543dfda2f80ee67f739c839d5039373ae821e97d":{"kind":"REWORK","_number":14,"created":"2019-06-17 12:57:00.000000000","uploader":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"ref":"refs/changes/98/649298/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/14"}}},"commit":{"parents":[{"commit":"5bbdcba0dd8747f080831b9b8be374ff3f6426d9","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/5bbdcba0dd8747f080831b9b8be374ff3f6426d9"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","date":"2019-06-17 12:56:35.000000000","tz":120},"subject":"Create Docker image and build in Zuul for Monasca-Thresh","message":"Create Docker image and build in Zuul for Monasca-Thresh\n\nStory: 2001694\nTask: 30289\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/543dfda2f80ee67f739c839d5039373ae821e97d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/543dfda2f80ee67f739c839d5039373ae821e97d"}]},"branch":"refs/heads/master"},"74dc6277762ae38820027660a238cf8344a54327":{"kind":"REWORK","_number":15,"created":"2019-06-18 13:13:37.000000000","uploader":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"ref":"refs/changes/98/649298/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/15"}}},"commit":{"parents":[{"commit":"5bbdcba0dd8747f080831b9b8be374ff3f6426d9","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/5bbdcba0dd8747f080831b9b8be374ff3f6426d9"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","date":"2019-06-18 13:13:18.000000000","tz":120},"subject":"Create Docker image and build in Zuul for Monasca-Thresh","message":"Create Docker image and build in Zuul for Monasca-Thresh\n\nStory: 2001694\nTask: 30289\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/74dc6277762ae38820027660a238cf8344a54327"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/74dc6277762ae38820027660a238cf8344a54327"}]},"branch":"refs/heads/master"},"bdb81d75936ba00dba02033ce642e9c048119510":{"kind":"REWORK","_number":16,"created":"2019-07-01 12:38:48.000000000","uploader":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"ref":"refs/changes/98/649298/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/16"}}},"commit":{"parents":[{"commit":"5bbdcba0dd8747f080831b9b8be374ff3f6426d9","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/5bbdcba0dd8747f080831b9b8be374ff3f6426d9"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","date":"2019-07-01 12:38:18.000000000","tz":120},"subject":"Create Docker image and build in Zuul for Monasca-Thresh","message":"Create Docker image and build in Zuul for Monasca-Thresh\n\nStory: 2001694\nTask: 30289\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/bdb81d75936ba00dba02033ce642e9c048119510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/bdb81d75936ba00dba02033ce642e9c048119510"}]},"branch":"refs/heads/master"},"2a99e0331a6c75092065d8ade11c8e1e01d9e3c4":{"kind":"REWORK","_number":17,"created":"2019-07-09 09:36:24.000000000","uploader":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"ref":"refs/changes/98/649298/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/17"}}},"commit":{"parents":[{"commit":"5bbdcba0dd8747f080831b9b8be374ff3f6426d9","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/5bbdcba0dd8747f080831b9b8be374ff3f6426d9"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","date":"2019-07-09 09:36:11.000000000","tz":120},"subject":"Create Docker image and build in Zuul for Monasca-Thresh","message":"Create Docker image and build in Zuul for Monasca-Thresh\n\nStory: 2001694\nTask: 30289\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/2a99e0331a6c75092065d8ade11c8e1e01d9e3c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/2a99e0331a6c75092065d8ade11c8e1e01d9e3c4"}]},"branch":"refs/heads/master"},"88d2c05bfb0bf73c9ff8017b75ad9a58d2057328":{"kind":"REWORK","_number":18,"created":"2019-07-16 14:29:25.000000000","uploader":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"ref":"refs/changes/98/649298/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/18"}}},"commit":{"parents":[{"commit":"5bbdcba0dd8747f080831b9b8be374ff3f6426d9","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/5bbdcba0dd8747f080831b9b8be374ff3f6426d9"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","date":"2019-07-16 14:28:55.000000000","tz":120},"subject":"Create Docker image and build in Zuul for Monasca-Thresh","message":"Create Docker image and build in Zuul for Monasca-Thresh\n\nStory: 2001694\nTask: 30289\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/88d2c05bfb0bf73c9ff8017b75ad9a58d2057328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/88d2c05bfb0bf73c9ff8017b75ad9a58d2057328"}]},"branch":"refs/heads/master"},"c3268b7ec75305ae704aaed785af8aabd1973367":{"kind":"REWORK","_number":19,"created":"2019-07-25 10:19:47.000000000","uploader":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"ref":"refs/changes/98/649298/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/19"}}},"commit":{"parents":[{"commit":"5bbdcba0dd8747f080831b9b8be374ff3f6426d9","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/5bbdcba0dd8747f080831b9b8be374ff3f6426d9"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","date":"2019-07-25 10:19:20.000000000","tz":120},"subject":"Create Docker image and build in Zuul for Monasca-Thresh","message":"Create Docker image and build in Zuul for Monasca-Thresh\n\nStory: 2001694\nTask: 30289\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/c3268b7ec75305ae704aaed785af8aabd1973367"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/c3268b7ec75305ae704aaed785af8aabd1973367"}]},"branch":"refs/heads/master"},"94f97ca2cd49256e7472e9f10567d7512025e7eb":{"kind":"REWORK","_number":20,"created":"2019-07-25 14:10:34.000000000","uploader":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"ref":"refs/changes/98/649298/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/20"}}},"commit":{"parents":[{"commit":"5bbdcba0dd8747f080831b9b8be374ff3f6426d9","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/5bbdcba0dd8747f080831b9b8be374ff3f6426d9"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","date":"2019-07-25 14:10:22.000000000","tz":120},"subject":"Create Docker image and build in Zuul for Monasca-Thresh","message":"Create Docker image and build in Zuul for Monasca-Thresh\n\nStory: 2001694\nTask: 30289\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/94f97ca2cd49256e7472e9f10567d7512025e7eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/94f97ca2cd49256e7472e9f10567d7512025e7eb"}]},"branch":"refs/heads/master"},"7bfb8407f343078a08e261ce61135e4e1b0f810b":{"kind":"REWORK","_number":21,"created":"2019-08-05 14:00:17.000000000","uploader":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"ref":"refs/changes/98/649298/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/21"}}},"commit":{"parents":[{"commit":"5bbdcba0dd8747f080831b9b8be374ff3f6426d9","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/5bbdcba0dd8747f080831b9b8be374ff3f6426d9"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","date":"2019-08-05 14:00:08.000000000","tz":120},"subject":"Create Docker image and build in Zuul for Monasca-Thresh","message":"Create Docker image and build in Zuul for Monasca-Thresh\n\nStory: 2001694\nTask: 30289\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/7bfb8407f343078a08e261ce61135e4e1b0f810b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/7bfb8407f343078a08e261ce61135e4e1b0f810b"}]},"branch":"refs/heads/master"},"4e337adce15d437a2e178725cd770bbdc88cf020":{"kind":"REWORK","_number":22,"created":"2019-08-06 13:07:24.000000000","uploader":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"ref":"refs/changes/98/649298/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/22"}}},"commit":{"parents":[{"commit":"5bbdcba0dd8747f080831b9b8be374ff3f6426d9","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/5bbdcba0dd8747f080831b9b8be374ff3f6426d9"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","date":"2019-08-06 13:07:14.000000000","tz":120},"subject":"Create Docker image and build in Zuul for Monasca-Thresh","message":"Create Docker image and build in Zuul for Monasca-Thresh\n\nStory: 2001694\nTask: 30289\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/4e337adce15d437a2e178725cd770bbdc88cf020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/4e337adce15d437a2e178725cd770bbdc88cf020"}]},"branch":"refs/heads/master"},"5cb9f94387dcbd76a7d261b047b65aa63aeb5092":{"kind":"REWORK","_number":23,"created":"2019-08-29 08:09:26.000000000","uploader":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"ref":"refs/changes/98/649298/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/98/649298/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/98/649298/23"}}},"commit":{"parents":[{"commit":"5bbdcba0dd8747f080831b9b8be374ff3f6426d9","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/5bbdcba0dd8747f080831b9b8be374ff3f6426d9"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-04-02 12:25:12.000000000","tz":120},"committer":{"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","date":"2019-08-29 08:09:14.000000000","tz":120},"subject":"Create Docker image and build in Zuul for Monasca-Thresh","message":"Create Docker image and build in Zuul for Monasca-Thresh\n\nStory: 2001694\nTask: 30289\n\nChange-Id: If3323a18ac3697a92ca3f06b0db154517582d7aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/5cb9f94387dcbd76a7d261b047b65aa63aeb5092"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/5cb9f94387dcbd76a7d261b047b65aa63aeb5092"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
