)]}'
{"id":"openstack%2Fmonasca-thresh~675872","triplet_id":"openstack%2Fmonasca-thresh~master~I0717cc775fa1d7b38d14ade5c51996740d1b7e26","project":"openstack/monasca-thresh","branch":"master","topic":"docker","hashtags":[],"change_id":"I0717cc775fa1d7b38d14ade5c51996740d1b7e26","subject":"Push Docker image to Docker Hub from Zuul","status":"MERGED","created":"2019-08-12 11:50:03.000000000","updated":"2019-09-25 10:06:11.000000000","submitted":"2019-09-25 10:06:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675872-1569405971901-b42f53c0","meta_rev_id":"52e8fece9f5dea0a36dcbccc980d23afcc6f90ab","_number":675872,"virtual_id_number":675872,"owner":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 10:06:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-25 09:59:49.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-25 09:59:49.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 09:59:49.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2019-09-25 10:06:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"696f75aa3215e232ab1346b4d6b621b9c2fd206a","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-08-12 11:50:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d0ac312a4067eb9bb7852b8fe419e4686159c4b","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-08-12 11:53:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fefbb8c446111420848ac51d51540ffa8e978187","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 12:45:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb https://logs.opendev.org/72/675872/2/check/monasca-tempest-python3-influxdb/f2e1781/ : SUCCESS in 45m 59s (non-voting)\n- monasca-tempest-python3-cassandra https://logs.opendev.org/72/675872/2/check/monasca-tempest-python3-cassandra/9b8c55b/ : FAILURE in 24m 08s (non-voting)\n- monasca-tempest-java-influxdb https://logs.opendev.org/72/675872/2/check/monasca-tempest-java-influxdb/d915032/ : FAILURE in 46m 29s (non-voting)\n- legacy-monasca-thresh-maven-build https://logs.opendev.org/72/675872/2/check/legacy-monasca-thresh-maven-build/566baf3/ : SUCCESS in 7m 12s\n- build-monasca-docker-image https://logs.opendev.org/72/675872/2/check/build-monasca-docker-image/a7a236d/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":2},{"id":"ab2b54d12257660e6d065fd5b3cce0b633ad5452","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-09-25 09:59:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8cc2ba5bb7e8e37aaa9b5bd6e035ed33d8301b42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 10:00:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"17d440bd20c4e556c662cea50551d1e061f0203a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 10:06:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/c654d438198047dc93e8634a12ec3fb5 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"66d3f715dd3ce0c1d6ac3a5e9af5f2b54268660b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 10:06:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"78c66d7f2e02d642bac2dba8e793e39230cd93ad","revisions":{"be4f111eed2d1938cf260da99056c14af91c9f0f":{"kind":"REWORK","_number":1,"created":"2019-08-12 11:50:03.000000000","uploader":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"ref":"refs/changes/72/675872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/72/675872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/72/675872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/72/675872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/72/675872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/72/675872/1"}}},"commit":{"parents":[{"commit":"b1409c391f72826baac7951a753a58672874d88f","subject":"Merge \"Enable monasca-thresh to set topology.max.spout.pending f. Storm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/b1409c391f72826baac7951a753a58672874d88f"}]}],"author":{"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","date":"2019-08-12 11:45:46.000000000","tz":120},"committer":{"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","date":"2019-08-12 11:45:46.000000000","tz":120},"subject":"Push Docker image to Docker Hub from Zuul","message":"Push Docker image to Docker Hub from Zuul\n\nStory: 2001694\nTask: 36260\n\nChange-Id: I0717cc775fa1d7b38d14ade5c51996740d1b7e26\nDepends-On: https://review.opendev.org/649298/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/be4f111eed2d1938cf260da99056c14af91c9f0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/be4f111eed2d1938cf260da99056c14af91c9f0f"}]},"branch":"refs/heads/master"},"78c66d7f2e02d642bac2dba8e793e39230cd93ad":{"kind":"REWORK","_number":2,"created":"2019-08-12 11:53:14.000000000","uploader":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"ref":"refs/changes/72/675872/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/72/675872/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/72/675872/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/72/675872/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/72/675872/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/72/675872/2"}}},"commit":{"parents":[{"commit":"b1409c391f72826baac7951a753a58672874d88f","subject":"Merge \"Enable monasca-thresh to set topology.max.spout.pending f. Storm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/b1409c391f72826baac7951a753a58672874d88f"}]}],"author":{"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","date":"2019-08-12 11:45:46.000000000","tz":120},"committer":{"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","date":"2019-08-12 11:52:59.000000000","tz":120},"subject":"Push Docker image to Docker Hub from Zuul","message":"Push Docker image to Docker Hub from Zuul\n\nStory: 2001694\nTask: 36260\n\nChange-Id: I0717cc775fa1d7b38d14ade5c51996740d1b7e26\nDepends-On: https://review.opendev.org/649298/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/78c66d7f2e02d642bac2dba8e793e39230cd93ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/78c66d7f2e02d642bac2dba8e793e39230cd93ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
