)]}'
{"id":"openstack%2Fmonasca-thresh~729285","triplet_id":"openstack%2Fmonasca-thresh~master~I66292e20d54334e31f5240b170bb98751bce6777","project":"openstack/monasca-thresh","branch":"master","topic":"project-ptl-and-contrib-docs","hashtags":[],"change_id":"I66292e20d54334e31f5240b170bb98751bce6777","subject":"Add CONTRIBUTING.rst","status":"MERGED","created":"2020-05-19 14:27:46.000000000","updated":"2020-05-19 20:04:36.000000000","submitted":"2020-05-19 20:04:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729285-1589918676722-9d4ddafc","meta_rev_id":"a43e1cd3bc67b2f9af365852e10907c0e918645f","_number":729285,"virtual_id_number":729285,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-19 20:04:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-05-19 19:56:21.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-05-19 19:56:21.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 19:56:21.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-05-19 20:04:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"102db2ebac51c4fdfe0f122027c11e6561a7647d","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-05-19 14:27:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89b49155b9d93016ea6aa7b198c857d312089d98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 14:40:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/61c921d9b94c4bd99bd12d9e0a254fff : SUCCESS in 8m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ab5cb7e1084b106f21e29c8c3a6a1266b22ab551","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-05-19 14:53:02.000000000","message":"Topic set to project-ptl-and-contrib-docs","accounts_in_message":[],"_revision_number":1},{"id":"e8cdddc65588bcfef5e146bbcde3f90ecb7f95c0","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-05-19 19:56:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c098bae5c5f9c005393fc3aabcd8ffc4e7db005","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 19:56:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8c7e69a25241bb6974afdcc5901cbed6eb90c0fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 20:04:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a43e1cd3bc67b2f9af365852e10907c0e918645f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 20:04:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/28bad981129141b1afd12dbaa9ceedc8 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"846e255d782d40f27f786013a5a54ef5a5d96582","revisions":{"846e255d782d40f27f786013a5a54ef5a5d96582":{"kind":"REWORK","_number":1,"created":"2020-05-19 14:27:46.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/85/729285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/85/729285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/85/729285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/85/729285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/85/729285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/85/729285/1"}}},"commit":{"parents":[{"commit":"d400f7a2bb41aac6bd1a64add1565148e5439535","subject":"Merge \"Push Docker image to Docker Hub from Zuul\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/d400f7a2bb41aac6bd1a64add1565148e5439535"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-05-19 14:27:30.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-05-19 14:27:30.000000000","tz":120},"subject":"Add CONTRIBUTING.rst","message":"Add CONTRIBUTING.rst\n\nChange-Id: I66292e20d54334e31f5240b170bb98751bce6777\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/846e255d782d40f27f786013a5a54ef5a5d96582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/846e255d782d40f27f786013a5a54ef5a5d96582"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
