)]}'
{"id":"openstack%2Fmonasca-thresh~755918","triplet_id":"openstack%2Fmonasca-thresh~master~If23d9d3cf1daf4496cef87c2c4241502287901ba","project":"openstack/monasca-thresh","branch":"master","topic":"optimization/pip-no-cache-dir","hashtags":[],"change_id":"If23d9d3cf1daf4496cef87c2c4241502287901ba","subject":"optimize size and time using --no-cache-dir","status":"ABANDONED","created":"2020-10-03 06:34:09.000000000","updated":"2020-10-16 12:18:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a8b299bf755ad485bdcef23f37913726b37b0d6c","_number":755918,"virtual_id_number":755918,"owner":{"_account_id":31020,"name":"Pratik raj","email":"Rajpratik71@gmail.com","username":"rajpratik71"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-15 13:54:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2020-10-16 12:17:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-15 13:54:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-16 12:17:17.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"}],"messages":[{"id":"9e923f0627a9afb04bc5064a629814d6ee4a577e","author":{"_account_id":31020,"name":"Pratik raj","email":"Rajpratik71@gmail.com","username":"rajpratik71"},"date":"2020-10-03 06:34:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88fe39d7d05d798dd5892c4ad9297b3d6da2a732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 07:18:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/96460ab958764a7195dcce629e68c5b8 : FAILURE in 36m 10s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/e8b20212083b456eac84cc9f12705bed : FAILURE in 33m 57s (non-voting)\n- monasca-tempest-java-influxdb https://zuul.opendev.org/t/openstack/build/05b347eec9e7428eb2c15e7793a5941d : FAILURE in 37m 43s (non-voting)\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/25f0aab6c72c44baa13108423159d812 : SUCCESS in 6m 47s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/3748dcfdd71a4ae79b953ae3b56a4455 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4a63f8a53dd716a5a86e0b8960b45e05235241a6","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-10-15 12:55:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4f5c2f37783208a82bc5ddbe8bc336a22a7e453d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 13:54:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/0b59b6540c3340329f70b358da5d99af : SUCCESS in 42m 59s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/8bd8814805ca4c44a49702c0648cb91c : FAILURE in 35m 42s (non-voting)\n- monasca-tempest-java-influxdb https://zuul.opendev.org/t/openstack/build/8713ddf96107460b83de12753de9fa2a : FAILURE in 47m 39s (non-voting)\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/1e7e76b7b9a44091b2750b58e0dd59f3 : SUCCESS in 8m 04s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/eb620dce8eea40cdb6446f785c34893a : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f6afcd83d5dbf7162571f94b7d0aad3f04830e8d","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-10-16 12:17:17.000000000","message":"Patch Set 2: Code-Review-2\n\nRejected, please read about the reasons [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015515.html","accounts_in_message":[],"_revision_number":2},{"id":"02af1073a3b1bde0b9f647320b03005e1bd2675e","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-10-16 12:18:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"11078fdd89898056ca031a2d1603602d0d3b1b3e","revisions":{"1abb94586e293a7d4cc50991b0e175ab17a71e67":{"kind":"REWORK","_number":1,"created":"2020-10-03 06:34:09.000000000","uploader":{"_account_id":31020,"name":"Pratik raj","email":"Rajpratik71@gmail.com","username":"rajpratik71"},"ref":"refs/changes/18/755918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/18/755918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/18/755918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/18/755918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/18/755918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/18/755918/1"}}},"commit":{"parents":[{"commit":"846e255d782d40f27f786013a5a54ef5a5d96582","subject":"Add CONTRIBUTING.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/846e255d782d40f27f786013a5a54ef5a5d96582"}]}],"author":{"name":"Pratik Raj","email":"rajpratik71@gmail.com","date":"2020-10-03 06:33:38.000000000","tz":330},"committer":{"name":"Pratik Raj","email":"rajpratik71@gmail.com","date":"2020-10-03 06:34:02.000000000","tz":330},"subject":"optimize size and time using --no-cache-dir","message":"optimize size and time using --no-cache-dir\n\nUsing --no-cache-dir flag in pip install ,make sure dowloaded packages\nby pip don\u0027t cached on system . This is a best practise which make sure\nto fetch ftom repo instead of using local cached one . Further , in case\nof Docker Containers , by restricing caching , we can reduce image size.\nIn term of stats , it depends upon the number of python packages\nmultiplied by their respective size . e.g for heavy packages with a lot\nof dependencies it reduce a lot by don\u0027t caching pip packages.\n\nFurther , more detail information can be found at\n\nhttps://medium.com/sciforce/strategies-of-docker-images-optimization-2ca9cc5719b6\n\nChange-Id: If23d9d3cf1daf4496cef87c2c4241502287901ba\nSigned-off-by: Pratik Raj \u003crajpratik71@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/1abb94586e293a7d4cc50991b0e175ab17a71e67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/1abb94586e293a7d4cc50991b0e175ab17a71e67"}]},"branch":"refs/heads/master"},"11078fdd89898056ca031a2d1603602d0d3b1b3e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-15 12:55:27.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/18/755918/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/18/755918/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/18/755918/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/18/755918/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/18/755918/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/18/755918/2"}}},"commit":{"parents":[{"commit":"0ee0e79a73a2f5061473e8546638b1b5a3e0cde8","subject":"Add mariadb-connector-j and bump version to 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/0ee0e79a73a2f5061473e8546638b1b5a3e0cde8"}]}],"author":{"name":"Pratik Raj","email":"rajpratik71@gmail.com","date":"2020-10-03 06:33:38.000000000","tz":330},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2020-10-15 12:55:27.000000000","tz":0},"subject":"optimize size and time using --no-cache-dir","message":"optimize size and time using --no-cache-dir\n\nUsing --no-cache-dir flag in pip install ,make sure dowloaded packages\nby pip don\u0027t cached on system . This is a best practise which make sure\nto fetch ftom repo instead of using local cached one . Further , in case\nof Docker Containers , by restricing caching , we can reduce image size.\nIn term of stats , it depends upon the number of python packages\nmultiplied by their respective size . e.g for heavy packages with a lot\nof dependencies it reduce a lot by don\u0027t caching pip packages.\n\nFurther , more detail information can be found at\n\nhttps://medium.com/sciforce/strategies-of-docker-images-optimization-2ca9cc5719b6\n\nChange-Id: If23d9d3cf1daf4496cef87c2c4241502287901ba\nSigned-off-by: Pratik Raj \u003crajpratik71@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/11078fdd89898056ca031a2d1603602d0d3b1b3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/11078fdd89898056ca031a2d1603602d0d3b1b3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
