)]}'
{"id":"openstack%2Fmonasca-thresh~773928","triplet_id":"openstack%2Fmonasca-thresh~master~I3a3a51f50ad64ab1bd73672b0bc7b09884309e68","project":"openstack/monasca-thresh","branch":"master","topic":"Ensure-image-uses-storm-yaml","hashtags":[],"change_id":"I3a3a51f50ad64ab1bd73672b0bc7b09884309e68","subject":"Enable usage of storm.yaml","status":"MERGED","created":"2021-02-03 15:23:07.000000000","updated":"2021-02-11 13:12:58.000000000","submitted":"2021-02-11 13:12:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773928-Ensure-image-uses-storm-yaml","meta_rev_id":"7cf224a46d78b77dbc04d67727fc15977cfc4b74","_number":773928,"virtual_id_number":773928,"owner":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-11 13:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-11 13:03:49.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-11 13:03:49.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-03 15:24:44.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2021-02-03 16:21:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-11 11:46:20.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"}],"messages":[{"id":"14ce0a24aec1938da54ee2b2c18b2a372d63260c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"date":"2021-02-03 15:23:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fe113ac8398d8a0f59674cdbcdc2b698eaa67e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 16:21:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/efb5295a955d4b3c8a6551604687bb2c : SUCCESS in 38m 56s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/c3b5bf2fbfcd4e97abffeb45be026ae0 : FAILURE in 41m 12s (non-voting)\n- monasca-tempest-java-influxdb https://zuul.opendev.org/t/openstack/build/48e4cc690fd2465193f0669dd218d073 : FAILURE in 51m 21s (non-voting)\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/cfefc8b13d6f4d6695b88f375f10db70 : SUCCESS in 7m 35s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/4af3f8e3747f432db8e1656210f87935 : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":1},{"id":"53872eaed86275ada853c5d8468d87b3eb0dac8d","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2021-02-11 13:03:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3115c212c6fbf647579cca92daae14617dde14c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 13:04:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62d9dbfbdc777e6894510d9a9217c56442f06466","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 13:12:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/8bc1c1813a254e978807b38325ad0012 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7cf224a46d78b77dbc04d67727fc15977cfc4b74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 13:12:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1464b7b541b417c612ae34e78ed7363bbb8edc00","revisions":{"1464b7b541b417c612ae34e78ed7363bbb8edc00":{"kind":"REWORK","_number":1,"created":"2021-02-03 15:23:07.000000000","uploader":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"ref":"refs/changes/28/773928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/28/773928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/28/773928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/28/773928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/28/773928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/28/773928/1"}}},"commit":{"parents":[{"commit":"0ee0e79a73a2f5061473e8546638b1b5a3e0cde8","subject":"Add mariadb-connector-j and bump version to 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/0ee0e79a73a2f5061473e8546638b1b5a3e0cde8"}]}],"author":{"name":"bandorf","email":"Matthias.Bandorf@est.fujitsu.com","date":"2021-02-03 15:19:30.000000000","tz":60},"committer":{"name":"bandorf","email":"Matthias.Bandorf@est.fujitsu.com","date":"2021-02-03 15:19:30.000000000","tz":60},"subject":"Enable usage of storm.yaml","message":"Enable usage of storm.yaml\n\nStory: 2008592\nTask: 41749\nChange-Id: I3a3a51f50ad64ab1bd73672b0bc7b09884309e68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/1464b7b541b417c612ae34e78ed7363bbb8edc00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/1464b7b541b417c612ae34e78ed7363bbb8edc00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
