)]}'
{"id":"openstack%2Fmonasca-thresh~775667","triplet_id":"openstack%2Fmonasca-thresh~master~I5ac243ad6fa0755ddbc663177648c4ed7031613f","project":"openstack/monasca-thresh","branch":"master","topic":"eliminate-storm-warnings","hashtags":[],"change_id":"I5ac243ad6fa0755ddbc663177648c4ed7031613f","subject":"Change storm config to eliminate storm warnings","status":"MERGED","created":"2021-02-15 15:24:13.000000000","updated":"2021-02-16 12:46:35.000000000","submitted":"2021-02-16 12:46:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775667-eliminate-storm-warnings","meta_rev_id":"31f5c57d8845d2df6395051302fcb49700d509c8","_number":775667,"virtual_id_number":775667,"owner":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-16 12:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-16 12:37:19.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-16 12:37:19.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-15 16:19:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:19.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"}],"messages":[{"id":"bf6ba808eee7f7bae067665e21941808570a408e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"date":"2021-02-15 15:24:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4257a9122ae675a06b244294e429baf434390612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 16:19:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/4095ec8a49b5438db61dbd39c95f540e : SUCCESS in 42m 55s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/f75918136c1c4827b80963da8c1186d9 : FAILURE in 27m 29s (non-voting)\n- monasca-tempest-java-influxdb https://zuul.opendev.org/t/openstack/build/9be6ef463a234f078b35f640bf6e69d5 : FAILURE in 42m 19s (non-voting)\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/d8535bc7ce3440e692b579bda7145283 : SUCCESS in 6m 10s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/42623c4b57304c2fb359cf41c43f86e4 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"136c824c063a96240902ce25553919dbc3790ddb","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2021-02-16 12:37:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be747ea1f14ecdc791602cfe9a9f22dac7973a2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 12:39:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b9b30b28a169948b1a6988053e02d5c50963196","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 12:46:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/eb78a4586369425eaa23a7081f9f0a96 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"31f5c57d8845d2df6395051302fcb49700d509c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 12:46:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05dc74a9aea96b2eb732e3e3ff51f13d10c30aeb","revisions":{"05dc74a9aea96b2eb732e3e3ff51f13d10c30aeb":{"kind":"REWORK","_number":1,"created":"2021-02-15 15:24:13.000000000","uploader":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"ref":"refs/changes/67/775667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/67/775667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/67/775667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/67/775667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/67/775667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/67/775667/1"}}},"commit":{"parents":[{"commit":"67b9ab3a3035e031a3b8c72c6789d7cb9bf9b6a9","subject":"Merge \"Enable usage of storm.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/67b9ab3a3035e031a3b8c72c6789d7cb9bf9b6a9"}]}],"author":{"name":"bandorf","email":"Matthias.Bandorf@est.fujitsu.com","date":"2021-02-15 15:18:46.000000000","tz":60},"committer":{"name":"bandorf","email":"Matthias.Bandorf@est.fujitsu.com","date":"2021-02-15 15:18:46.000000000","tz":60},"subject":"Change storm config to eliminate storm warnings","message":"Change storm config to eliminate storm warnings\n\nStory: 2005483\nTask: 30576\nChange-Id: I5ac243ad6fa0755ddbc663177648c4ed7031613f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/05dc74a9aea96b2eb732e3e3ff51f13d10c30aeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/05dc74a9aea96b2eb732e3e3ff51f13d10c30aeb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
