)]}'
{"id":"openstack%2Fmonasca-thresh~802194","triplet_id":"openstack%2Fmonasca-thresh~stable%2Ftrain~I5d95763c3c28d8303ad19898e985ea89dab6fb76","project":"openstack/monasca-thresh","branch":"stable/train","topic":"zuul-remove-periodic-job","hashtags":[],"change_id":"I5d95763c3c28d8303ad19898e985ea89dab6fb76","subject":"Remove periodic docker publish job","status":"MERGED","created":"2021-07-25 17:30:07.000000000","updated":"2021-09-14 13:28:30.000000000","submitted":"2021-09-14 13:28:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802194-zuul-remove-periodic-job","meta_rev_id":"4ac304b75ec50caed660753aa92750df7a8265a0","_number":802194,"virtual_id_number":802194,"owner":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-14 13:28:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-14 13:18:14.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-14 13:18:14.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-25 18:12:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-14 13:18:14.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"}],"messages":[{"id":"a576603b43f0d87a53acd54ae49318f0fa9c25ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2021-07-25 17:30:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"568cb83bbb58b5d52678b7a613ffcffaabe3dbe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-25 18:12:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/f872d91a95784749b6f0c5e8133525db : FAILURE in 23m 53s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/ff4adbb4fe7a45778b293bc74a5f7a46 : FAILURE in 12m 41s (non-voting)\n- monasca-tempest-java-influxdb https://zuul.opendev.org/t/openstack/build/34447a822d134acab040177b41866ec4 : FAILURE in 40m 26s (non-voting)\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/f7de7cb7fbab42d8a071c51f9f141e34 : SUCCESS in 6m 17s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/e3abda92c44d471db0ae945109fff6f0 : FAILURE in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0f815a65339baf6013303615a6b7c2510c298b0e","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2021-08-31 13:13:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a2d28995fffd3def0e5e841ba29427f5aa2b7ac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 14:02:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/fdb89388f12d4b3183cdd83e89486529 : FAILURE in 22m 56s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/c8e35504751e44269b0a8f299f4c9995 : FAILURE in 13m 31s (non-voting)\n- monasca-tempest-java-influxdb https://zuul.opendev.org/t/openstack/build/959c3c5b280d4a4fb86b7ef4758e2b51 : FAILURE in 46m 37s (non-voting)\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/39eed6f0a6e94edfb72e34e3086bbcd8 : SUCCESS in 7m 18s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/0bc56e343210465f92bf54c4005fc654 : FAILURE in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"09aef4affb0f7b69545901b4114323edd3d8f634","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2021-09-13 13:10:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4ce9784098fe0fa72d1eba26bf644c8d38d2e80f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 14:02:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/242f017a31ad4df68dbe8f462a80ec7d : FAILURE in 28m 48s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/e6f7f1b42bbe42af9d3b980a21fc10a0 : FAILURE in 21m 52s (non-voting)\n- monasca-tempest-java-influxdb https://zuul.opendev.org/t/openstack/build/769ce3fd77bf48c58ca8843f2f72d246 : FAILURE in 47m 04s (non-voting)\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/3d064c6d6b9c42efa8fb94cde0c05925 : SUCCESS in 6m 12s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/665ca83b4ca3467ab506e37f65ac1740 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"fd666880e7b52d1ca994e67d27d9e0254bcff7d7","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-09-14 13:18:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"984f66c60f6769f5e77bbf98ac923aa76ffb2b68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 13:18:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4de4d6525eef0d797051b81d95dd0b00a9b625d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 13:28:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/01c8e91252af4daaa57939dba01e0303 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":2},{"id":"4ac304b75ec50caed660753aa92750df7a8265a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 13:28:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e4e6efc9f96d01afc693a9924d9c322d2b9b3d90","revisions":{"7f473d85558f6278060b3581b6d85ee8b60c4eeb":{"kind":"REWORK","_number":1,"created":"2021-07-25 17:30:07.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/94/802194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/94/802194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/94/802194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/94/802194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/94/802194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/94/802194/1"}}},"commit":{"parents":[{"commit":"4c2858d1f10f54ec3cdcc3367c4388ecdf0172f6","subject":"Fix zuul publish docker image job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/4c2858d1f10f54ec3cdcc3367c4388ecdf0172f6"}]}],"author":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2021-07-27 17:27:29.000000000","tz":-420},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2021-07-27 17:27:38.000000000","tz":-420},"subject":"Remove periodic docker publish job","message":"Remove periodic docker publish job\n\nHas been decided that we no longer need to publish docker image base on train every day\n\nChange-Id: I5d95763c3c28d8303ad19898e985ea89dab6fb76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/7f473d85558f6278060b3581b6d85ee8b60c4eeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/7f473d85558f6278060b3581b6d85ee8b60c4eeb"}]},"branch":"refs/heads/stable/train"},"e4e6efc9f96d01afc693a9924d9c322d2b9b3d90":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-09-13 13:10:33.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/94/802194/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/94/802194/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/94/802194/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/94/802194/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/94/802194/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/94/802194/2"}}},"commit":{"parents":[{"commit":"70ff0c13c1ac96a4112956c0da684a4cbae0847a","subject":"Fix docker build script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/70ff0c13c1ac96a4112956c0da684a4cbae0847a"}]}],"author":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2021-07-27 17:27:29.000000000","tz":-420},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2021-09-13 13:10:33.000000000","tz":0},"subject":"Remove periodic docker publish job","message":"Remove periodic docker publish job\n\nHas been decided that we no longer need to publish docker image base on train every day\n\nChange-Id: I5d95763c3c28d8303ad19898e985ea89dab6fb76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/e4e6efc9f96d01afc693a9924d9c322d2b9b3d90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/e4e6efc9f96d01afc693a9924d9c322d2b9b3d90"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
