)]}'
{"id":"openstack%2Fmonasca-thresh~923377","triplet_id":"openstack%2Fmonasca-thresh~master~I0e591770e30c321bcc340b9dd13303259462f30a","project":"openstack/monasca-thresh","branch":"master","topic":"zuul-config-erros","attention_set":{},"removed_from_attention_set":{"13114":{"account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"last_update":"2024-07-04 06:46:05.000000000","reason":"\u003cGERRIT_ACCOUNT_13114\u003e replied on the change","reason_account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}}},"hashtags":[],"change_id":"I0e591770e30c321bcc340b9dd13303259462f30a","subject":"Fix deprecated zuul queue syntax","status":"MERGED","created":"2024-07-03 10:00:19.000000000","updated":"2024-07-04 06:51:27.000000000","submitted":"2024-07-04 06:51:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923377-zuul-config-erros","meta_rev_id":"dea9e9e65a79259e2900a1c5bbe04acd4145b561","_number":923377,"virtual_id_number":923377,"owner":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 06:51:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":2,"date":"2024-07-04 06:46:05.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":1,"date":"2024-07-04 06:46:05.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-03 11:08:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37a59d79ec4e429e09c7fd3cd386374a3600d203","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-03 10:00:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94557766eaea06e1cdcc8983e4f24eb4322bf89e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 11:08:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2b46c02bdd249d9838ea1ef17a9059c\n\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/d0e257d727914e3b999774b0dd15a6af : SUCCESS in 57m 37s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/9898192a1cf5499b8574cfda281f0ac5 : FAILURE in 45m 25s (non-voting)\n- monasca-tempest-java-influxdb https://zuul.opendev.org/t/openstack/build/f4e025a6aae7419285a0e852df3e99e0 : FAILURE in 57m 35s (non-voting)\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/cdf06af96f204977b9cd62397321b365 : SUCCESS in 4m 57s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/3304b3af8daf4980a03b538f716e8873 : FAILURE in 2m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f1ddcfb877724c3e7a91df555ca117b708eda040","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-03 17:58:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e10573ad25ed79ca7f7bf31f7184117fc9065cd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-03 18:11:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"669834e3d8fde00da774b81f6a7528e660958fed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-03 18:57:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"958dff11458e85f4cad1b78d8cc23e274ce1c36d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-03 19:27:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"75fb81114065ea5ea21f04cd724e13e6935b5b25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 20:12:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2aeecbdf913340368e1edbc7dcd9bb11\n\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/80968799109543d78cdbb433885b108b : SUCCESS in 39m 47s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/de12438434d14e36ae75c2a9f298b597 : FAILURE in 39m 27s (non-voting)\n- monasca-tempest-java-influxdb https://zuul.opendev.org/t/openstack/build/e62251a7e2c34750b8516745a22e0775 : FAILURE in 37m 24s (non-voting)\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/750f94042c844b418be6e0a8814e6b44 : SUCCESS in 4m 35s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/ed1f5f5f1cd149e897d62966d920b19c : FAILURE in 2m 30s","accounts_in_message":[],"_revision_number":5},{"id":"6c0b190a484b58923e295ae12774994d789f5a44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-04 03:26:49.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"8de71905459f6b318d104064d29dfef1f91af67f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 04:14:17.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/58c38ef9a4be4ca3a93e908fafc37222\n\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/db7ea8a792f341bc8661e7b1b4fd3e73 : SUCCESS in 42m 26s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/acd36143350e48c28dfd2b75c45ab9de : FAILURE in 39m 16s (non-voting)\n- monasca-tempest-java-influxdb https://zuul.opendev.org/t/openstack/build/1378863955d54cbcbc8fce19d0cc0b69 : FAILURE in 38m 41s (non-voting)\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/d33e0b3aba434486ab79611d4c09a040 : SUCCESS in 8m 10s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/b971d18af3fa402c952937f8fc6c7d81 : FAILURE in 3m 13s","accounts_in_message":[],"_revision_number":6},{"id":"bce7016ba4ff47626ef87e2fd60c68e30b7086aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-04 05:50:28.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"96dd7ebbd67b71765b87665121e3013a30425899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 06:43:38.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56446a94457c491291f85527831e387e\n\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/3a6899acec1041d197dc206f2ed56c16 : SUCCESS in 52m 23s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/4d2deca3213e47c5afa5816c9e68cc08 : FAILURE in 43m 11s (non-voting)\n- monasca-tempest-java-influxdb https://zuul.opendev.org/t/openstack/build/0fdc0f4983c0464b85fa9b5b26463eee : FAILURE in 45m 59s (non-voting)\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/3596d04db5124fd88b8cd5aacfffbbb4 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":7},{"id":"6cfd1538f02e5ff054bd0549cb2278522c64ce5f","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-04 06:46:05.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1be4705a725a1483d311fc64aebccc5272fb9044","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 06:46:38.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"fbad7cf88b32f8d3d00893828292c6235c6b714b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 06:51:27.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df762dc0f5e2469080505b82ff76897f\n\n- legacy-monasca-thresh-maven-build https://zuul.opendev.org/t/openstack/build/3f88e72bc3f8459ba8c8a3afe0ba82ea : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":7},{"id":"dea9e9e65a79259e2900a1c5bbe04acd4145b561","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 06:51:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2d6a94c8e80946dae2441eef77667dcbabbfe058","revisions":{"3c6c5c82feb347c9ebdeea8bca078ea87e86b1b5":{"kind":"REWORK","_number":1,"created":"2024-07-03 10:00:19.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/77/923377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/77/923377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/1"}}},"commit":{"parents":[{"commit":"d256f05be6235dbaaff8ba738e6196b9c6241911","subject":"Merge \"Revert \"Use external Zookeeper in Local mode\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/d256f05be6235dbaaff8ba738e6196b9c6241911"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-03 09:59:36.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-03 10:00:15.000000000","tz":420},"subject":"Fix deprecated zuul queue syntax","message":"Fix deprecated zuul queue syntax\n\nZuul deprecated (and now removed) declaring shared queues at a pipeline\nlevel [1]. This patch fixes the syntax to make the gate work again for\nmonasca-common.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-May/028603.html\n[2] https://review.opendev.org/c/openstack/monasca-common/+/874297\n\nChange-Id: I0e591770e30c321bcc340b9dd13303259462f30a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/3c6c5c82feb347c9ebdeea8bca078ea87e86b1b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/3c6c5c82feb347c9ebdeea8bca078ea87e86b1b5"}]},"branch":"refs/heads/master"},"68d5dc665d1aa0ece0f63618f38bfec06dcbaec6":{"kind":"REWORK","_number":2,"created":"2024-07-03 17:58:11.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/77/923377/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/77/923377/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/2"}}},"commit":{"parents":[{"commit":"d256f05be6235dbaaff8ba738e6196b9c6241911","subject":"Merge \"Revert \"Use external Zookeeper in Local mode\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/d256f05be6235dbaaff8ba738e6196b9c6241911"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-03 09:59:36.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-03 17:57:58.000000000","tz":420},"subject":"Fix deprecated zuul queue syntax","message":"Fix deprecated zuul queue syntax\n\nZuul deprecated (and now removed) declaring shared queues at a pipeline\nlevel [1]. This patch fixes the syntax to make the gate work again for\nmonasca-common.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-May/028603.html\n[2] https://review.opendev.org/c/openstack/monasca-common/+/874297\n\nChange-Id: I0e591770e30c321bcc340b9dd13303259462f30a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/68d5dc665d1aa0ece0f63618f38bfec06dcbaec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/68d5dc665d1aa0ece0f63618f38bfec06dcbaec6"}]},"branch":"refs/heads/master"},"1f6fb6882743af95014e0152ff2382de0fe36d4e":{"kind":"REWORK","_number":3,"created":"2024-07-03 18:11:23.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/77/923377/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/77/923377/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/3"}}},"commit":{"parents":[{"commit":"d256f05be6235dbaaff8ba738e6196b9c6241911","subject":"Merge \"Revert \"Use external Zookeeper in Local mode\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/d256f05be6235dbaaff8ba738e6196b9c6241911"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-03 09:59:36.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-03 18:11:10.000000000","tz":420},"subject":"Fix deprecated zuul queue syntax","message":"Fix deprecated zuul queue syntax\n\nZuul deprecated (and now removed) declaring shared queues at a pipeline\nlevel [1]. This patch fixes the syntax to make the gate work again for\nmonasca-common.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-May/028603.html\n[2] https://review.opendev.org/c/openstack/monasca-common/+/874297\n\nChange-Id: I0e591770e30c321bcc340b9dd13303259462f30a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/1f6fb6882743af95014e0152ff2382de0fe36d4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/1f6fb6882743af95014e0152ff2382de0fe36d4e"}]},"branch":"refs/heads/master"},"4f3918ff43797e60ff3c26085f0b07d6ffce58c8":{"kind":"REWORK","_number":4,"created":"2024-07-03 18:57:31.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/77/923377/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/77/923377/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/4"}}},"commit":{"parents":[{"commit":"d256f05be6235dbaaff8ba738e6196b9c6241911","subject":"Merge \"Revert \"Use external Zookeeper in Local mode\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/d256f05be6235dbaaff8ba738e6196b9c6241911"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-03 09:59:36.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-03 18:57:18.000000000","tz":420},"subject":"Fix deprecated zuul queue syntax","message":"Fix deprecated zuul queue syntax\n\nZuul deprecated (and now removed) declaring shared queues at a pipeline\nlevel [1]. This patch fixes the syntax to make the gate work again for\nmonasca-common.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-May/028603.html\n[2] https://review.opendev.org/c/openstack/monasca-common/+/874297\n\nChange-Id: I0e591770e30c321bcc340b9dd13303259462f30a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/4f3918ff43797e60ff3c26085f0b07d6ffce58c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/4f3918ff43797e60ff3c26085f0b07d6ffce58c8"}]},"branch":"refs/heads/master"},"ad839cd486046e823873c2767e11190ed07ed6a5":{"kind":"REWORK","_number":5,"created":"2024-07-03 19:27:30.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/77/923377/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/77/923377/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/5"}}},"commit":{"parents":[{"commit":"d256f05be6235dbaaff8ba738e6196b9c6241911","subject":"Merge \"Revert \"Use external Zookeeper in Local mode\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/d256f05be6235dbaaff8ba738e6196b9c6241911"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-03 09:59:36.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-03 19:27:15.000000000","tz":420},"subject":"Fix deprecated zuul queue syntax","message":"Fix deprecated zuul queue syntax\n\nZuul deprecated (and now removed) declaring shared queues at a pipeline\nlevel [1]. This patch fixes the syntax to make the gate work again for\nmonasca-common.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-May/028603.html\n[2] https://review.opendev.org/c/openstack/monasca-common/+/874297\n\nChange-Id: I0e591770e30c321bcc340b9dd13303259462f30a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/ad839cd486046e823873c2767e11190ed07ed6a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/ad839cd486046e823873c2767e11190ed07ed6a5"}]},"branch":"refs/heads/master"},"442efa9cf9d9f944c871fd4895c5fa996bc429db":{"kind":"REWORK","_number":6,"created":"2024-07-04 03:26:49.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/77/923377/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/77/923377/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/6"}}},"commit":{"parents":[{"commit":"d256f05be6235dbaaff8ba738e6196b9c6241911","subject":"Merge \"Revert \"Use external Zookeeper in Local mode\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/d256f05be6235dbaaff8ba738e6196b9c6241911"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-03 09:59:36.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-04 03:26:36.000000000","tz":420},"subject":"Fix deprecated zuul queue syntax","message":"Fix deprecated zuul queue syntax\n\nZuul deprecated (and now removed) declaring shared queues at a pipeline\nlevel [1]. This patch fixes the syntax to make the gate work again for\nmonasca-common.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-May/028603.html\n[2] https://review.opendev.org/c/openstack/monasca-common/+/874297\n\nChange-Id: I0e591770e30c321bcc340b9dd13303259462f30a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/442efa9cf9d9f944c871fd4895c5fa996bc429db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/442efa9cf9d9f944c871fd4895c5fa996bc429db"}]},"branch":"refs/heads/master"},"2d6a94c8e80946dae2441eef77667dcbabbfe058":{"kind":"REWORK","_number":7,"created":"2024-07-04 05:50:28.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/77/923377/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-thresh","ref":"refs/changes/77/923377/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-thresh refs/changes/77/923377/7"}}},"commit":{"parents":[{"commit":"d256f05be6235dbaaff8ba738e6196b9c6241911","subject":"Merge \"Revert \"Use external Zookeeper in Local mode\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/d256f05be6235dbaaff8ba738e6196b9c6241911"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-03 09:59:36.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-04 05:50:06.000000000","tz":420},"subject":"Fix deprecated zuul queue syntax","message":"Fix deprecated zuul queue syntax\n\nZuul deprecated (and now removed) declaring shared queues at a pipeline\nlevel [1]. This patch fixes the syntax to make the gate work again for\nmonasca-common.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-May/028603.html\n[2] https://review.opendev.org/c/openstack/monasca-common/+/874297\n\nChange-Id: I0e591770e30c321bcc340b9dd13303259462f30a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/2d6a94c8e80946dae2441eef77667dcbabbfe058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-thresh/commit/2d6a94c8e80946dae2441eef77667dcbabbfe058"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
