)]}'
{"id":"openstack%2Fmonasca-transform~957081","triplet_id":"openstack%2Fmonasca-transform~master~I37a14765a10d0535a2ab782c37681c0bebab06ca","project":"openstack/monasca-transform","branch":"master","topic":"retire-monasca-transform","hashtags":["monasca-retirement"],"change_id":"I37a14765a10d0535a2ab782c37681c0bebab06ca","subject":"Retire monasca-transform repository","status":"ABANDONED","created":"2025-08-12 05:21:39.000000000","updated":"2025-08-15 18:32:24.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6050b65569155ba585e08567291646086e43d5c2","_number":957081,"virtual_id_number":957081,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-12 05:26:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":-1,"date":"2025-08-13 21:30:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-12 05:26:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a08e72d0efdda1d69513c77e96ec46960194735a","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-12 05:21:39.000000000","message":"Hashtag added: monasca-retirement","accounts_in_message":[],"_revision_number":1},{"id":"1a8a977b246329f690ca586e9db7aa00915fcbb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 05:26:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0ca0b56c65847989cbf1e0a48fb0524\n\n- noop https://zuul.opendev.org/t/openstack/build/64d9e18652014a1c815214e46f836292 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"51af8a8cae60263d9b16a120460957a3976fad21","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-13 14:39:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"affe2917d4690ca46ea04be3ae39e160a40e49b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 14:45:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa72455aba0d4c3b986c94e12a2808dd\n\n- noop https://zuul.opendev.org/t/openstack/build/a3e6499d18724976853b2a19cd4552ee : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"bd4c8f5d27ac8372331c57762ec9553ab6517e69","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-13 21:30:34.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"6050b65569155ba585e08567291646086e43d5c2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-15 18:32:24.000000000","message":"Abandoned\n\nNo plans to unretire this to merely update the readme. Will leave this as is.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87cb9f889c9c79660237f34546e940e90dd95d0c","revisions":{"87cb9f889c9c79660237f34546e940e90dd95d0c":{"kind":"REWORK","_number":1,"created":"2025-08-12 05:21:39.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/81/957081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-transform","ref":"refs/changes/81/957081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-transform refs/changes/81/957081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-transform refs/changes/81/957081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-transform refs/changes/81/957081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-transform refs/changes/81/957081/1"}}},"commit":{"parents":[{"commit":"ddf7e04d324cbc485cda5fae4e039070a8be894f","subject":"Moving IRC network reference to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-transform/commit/ddf7e04d324cbc485cda5fae4e039070a8be894f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-08-12 05:15:43.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-08-12 05:21:37.000000000","tz":-420},"subject":"Retire monasca-transform repository","message":"Retire monasca-transform repository\n\nThis repository is being retired as part of the Monasca project\nretirement. The project content has been replaced with a retirement\nnotice.\n\nDepends-On: I3cb522ce8f51424b64e93c1efaf0dfd1781cd5ac\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/957063\n\nChange-Id: I37a14765a10d0535a2ab782c37681c0bebab06ca\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-transform/commit/87cb9f889c9c79660237f34546e940e90dd95d0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-transform/commit/87cb9f889c9c79660237f34546e940e90dd95d0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
