)]}'
{"id":"openstack%2Fmonasca-ui~652419","triplet_id":"openstack%2Fmonasca-ui~master~I7ea4cb2da696cfa10551f4cab645db6d47de23d9","project":"openstack/monasca-ui","branch":"master","topic":"dropping-py35-testing","hashtags":[],"change_id":"I7ea4cb2da696cfa10551f4cab645db6d47de23d9","subject":"Dropping the py35 testing","status":"MERGED","created":"2019-04-14 23:56:40.000000000","updated":"2019-05-21 15:08:37.000000000","submitted":"2019-05-21 15:08:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652419-1558451317198-334d6ac5","meta_rev_id":"d53c1a5858146c75d6514ac2d33ca8c6baf723ba","_number":652419,"virtual_id_number":652419,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":2,"date":"2019-05-21 15:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"all":[{"value":2,"date":"2019-05-20 15:54:31.000000000","_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-21 14:55:39.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-21 14:55:39.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 15:54:31.000000000","updated_by":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"reviewer":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"state":"REVIEWER"},{"updated":"2019-05-21 14:55:39.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-05-21 15:08:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5efd2f220664d8c723382b30951fddc235264b75","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-14 23:56:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac0ab0e1fad5aee710d8fdccba0b44c89ef20a7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 00:05:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/19/652419/1/check/test-release-openstack/ee7778b/ : SUCCESS in 3m 23s\n- openstack-tox-cover http://logs.openstack.org/19/652419/1/check/openstack-tox-cover/effc1dc/ : FAILURE in 6m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/652419/1/check/openstack-tox-lower-constraints/44062f5/ : SUCCESS in 6m 12s\n- openstack-tox-pep8 http://logs.openstack.org/19/652419/1/check/openstack-tox-pep8/12f941b/ : SUCCESS in 5m 25s\n- openstack-tox-py27 http://logs.openstack.org/19/652419/1/check/openstack-tox-py27/260d808/ : FAILURE in 5m 27s\n- openstack-tox-py36 http://logs.openstack.org/19/652419/1/check/openstack-tox-py36/27dd157/ : FAILURE in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6246436566e0086118456e17ebda96d906eaee3c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-15 18:12:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"62f3853e37c0e6438658f7761064a88c8b55d9d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 18:40:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/19/652419/2/check/test-release-openstack/b1f90e7/ : SUCCESS in 5m 24s\n- openstack-tox-cover http://logs.openstack.org/19/652419/2/check/openstack-tox-cover/9041a88/ : FAILURE in 9m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/652419/2/check/openstack-tox-lower-constraints/1c9119b/ : SUCCESS in 6m 52s\n- openstack-tox-pep8 http://logs.openstack.org/19/652419/2/check/openstack-tox-pep8/94e7b5d/ : SUCCESS in 6m 51s\n- openstack-tox-py27 http://logs.openstack.org/19/652419/2/check/openstack-tox-py27/365a5a9/ : FAILURE in 7m 01s\n- openstack-tox-py36 http://logs.openstack.org/19/652419/2/check/openstack-tox-py36/9ebf87d/ : FAILURE in 7m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c61e9d2637f3bd13a968536acb6c49923a771869","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-05-15 15:36:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"33b0625101b3977f7a606452d04a90938266cdb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 16:09:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/19/652419/2/check/test-release-openstack/a23d2cf/ : SUCCESS in 3m 21s\n- openstack-tox-cover http://logs.openstack.org/19/652419/2/check/openstack-tox-cover/f4c08ca/ : FAILURE in 6m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/652419/2/check/openstack-tox-lower-constraints/88fb424/ : SUCCESS in 5m 57s\n- openstack-tox-pep8 http://logs.openstack.org/19/652419/2/check/openstack-tox-pep8/16fa05a/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/19/652419/2/check/openstack-tox-py27/ecd9e57/ : FAILURE in 5m 03s\n- openstack-tox-py36 http://logs.openstack.org/19/652419/2/check/openstack-tox-py36/fa24613/ : FAILURE in 5m 43s","accounts_in_message":[],"_revision_number":2},{"id":"04073c90db40c194e42fcb7c5ae8bce25579020c","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-05-20 15:07:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4c72eedad539fe2ed797c214f6c95ccf4818d227","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 15:21:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/19/652419/3/check/test-release-openstack/4329aa5/ : SUCCESS in 3m 26s\n- openstack-tox-cover http://logs.openstack.org/19/652419/3/check/openstack-tox-cover/e785e83/cover/ : SUCCESS in 6m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/652419/3/check/openstack-tox-lower-constraints/6686368/ : SUCCESS in 6m 14s\n- openstack-tox-pep8 http://logs.openstack.org/19/652419/3/check/openstack-tox-pep8/ab2e93c/ : SUCCESS in 5m 05s\n- openstack-tox-py27 http://logs.openstack.org/19/652419/3/check/openstack-tox-py27/2caa109/ : SUCCESS in 6m 29s\n- openstack-tox-py36 http://logs.openstack.org/19/652419/3/check/openstack-tox-py36/6e98add/ : SUCCESS in 9m 11s","accounts_in_message":[],"_revision_number":3},{"id":"b925371cdb3c987a1b0d45f9923e3a6d3c30bb7d","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-05-20 15:54:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"136d063f94ab8daca9bdcbd20052ff6780586260","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-05-20 15:54:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e0a44ee51738251524ef43af3d5a8bded1392073","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-05-21 14:55:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"042c70607403c72fe8786aac82f642dbfd81accf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 14:55:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9d93be4964f075a393ddd45ff983e1f3a9f12714","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 15:08:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/19/652419/3/gate/test-release-openstack/0c37281/ : SUCCESS in 3m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/652419/3/gate/openstack-tox-lower-constraints/cb13175/ : SUCCESS in 6m 37s\n- openstack-tox-pep8 http://logs.openstack.org/19/652419/3/gate/openstack-tox-pep8/1971303/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/19/652419/3/gate/openstack-tox-py27/ef49345/ : SUCCESS in 4m 59s\n- openstack-tox-py36 http://logs.openstack.org/19/652419/3/gate/openstack-tox-py36/f3f2c0a/ : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"8b84e35d009f6ba0bde6c6683ca55c1bf9f0c80f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 15:08:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"df89ad3048dac7aad2bc44dccc30cd2b543714a5","revisions":{"2fadbe7aa11742286e98105dcbc1be051139eb1a":{"kind":"REWORK","_number":1,"created":"2019-04-14 23:56:40.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/19/652419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/19/652419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/19/652419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/19/652419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/19/652419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/19/652419/1"}}},"commit":{"parents":[{"commit":"a6bc9ba3ced079d2f662c83ad86dfb9dfa8658df","subject":"Merge \"fix a typo in drilldown.js\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/a6bc9ba3ced079d2f662c83ad86dfb9dfa8658df"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-14 23:56:21.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-14 23:56:38.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I7ea4cb2da696cfa10551f4cab645db6d47de23d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/2fadbe7aa11742286e98105dcbc1be051139eb1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/2fadbe7aa11742286e98105dcbc1be051139eb1a"}]},"branch":"refs/heads/master"},"27eca91623326da86699a0ec91c32f9b8719556a":{"kind":"REWORK","_number":2,"created":"2019-04-15 18:12:53.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/19/652419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/19/652419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/19/652419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/19/652419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/19/652419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/19/652419/2"}}},"commit":{"parents":[{"commit":"a6bc9ba3ced079d2f662c83ad86dfb9dfa8658df","subject":"Merge \"fix a typo in drilldown.js\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/a6bc9ba3ced079d2f662c83ad86dfb9dfa8658df"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-14 23:56:21.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 18:12:49.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I7ea4cb2da696cfa10551f4cab645db6d47de23d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/27eca91623326da86699a0ec91c32f9b8719556a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/27eca91623326da86699a0ec91c32f9b8719556a"}]},"branch":"refs/heads/master"},"df89ad3048dac7aad2bc44dccc30cd2b543714a5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-20 15:07:59.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/19/652419/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/19/652419/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/19/652419/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/19/652419/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/19/652419/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/19/652419/3"}}},"commit":{"parents":[{"commit":"c85ecafe59a580042ac3860c67b3f1cc44dd5830","subject":"Merge \"Fix failing unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/c85ecafe59a580042ac3860c67b3f1cc44dd5830"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-14 23:56:21.000000000","tz":0},"committer":{"name":"Witold Bedyk","email":"witold.bedyk@suse.com","date":"2019-05-20 15:07:59.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I7ea4cb2da696cfa10551f4cab645db6d47de23d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/df89ad3048dac7aad2bc44dccc30cd2b543714a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/df89ad3048dac7aad2bc44dccc30cd2b543714a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
