)]}'
{"id":"openstack%2Fmonasca-ui~654845","triplet_id":"openstack%2Fmonasca-ui~master~I08632e63eef7cb81f0b3828db69feb9750d65cb5","project":"openstack/monasca-ui","branch":"master","hashtags":[],"change_id":"I08632e63eef7cb81f0b3828db69feb9750d65cb5","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 08:41:30.000000000","updated":"2019-05-20 15:17:47.000000000","submitted":"2019-05-20 15:17:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654845-1558365467907-33725622","meta_rev_id":"49fe554a5f2bc41724cdffe92b9932dc6aab7edc","_number":654845,"virtual_id_number":654845,"owner":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-05-20 15:17:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"all":[{"value":2,"date":"2019-05-20 14:17:00.000000000","_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":1,"date":"2019-05-20 14:17:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2019-05-20 14:17:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2019-05-20 14:17:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":1,"date":"2019-05-20 14:17:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":2,"date":"2019-05-20 15:04:48.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":1,"date":"2019-05-20 14:17:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":1,"date":"2019-05-20 15:04:48.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 11:11:53.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2019-04-24 17:35:54.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-05-08 12:50:33.000000000","updated_by":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"reviewer":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"state":"REVIEWER"},{"updated":"2019-05-09 01:57:40.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-05-13 13:46:20.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-05-14 02:57:02.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-05-14 08:07:40.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-05-20 15:04:48.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-05-20 15:17:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b71dcc2c3081065c443c3a7176957bda3d1371c6","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-23 08:41:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1b54ccfbff375eab9f89a0c3dff6a12bd82e3f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 09:06:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/45/654845/1/check/test-release-openstack/bcc3dfd/ : SUCCESS in 3m 43s\n- openstack-tox-cover http://logs.openstack.org/45/654845/1/check/openstack-tox-cover/6bef8ef/ : FAILURE in 6m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/654845/1/check/openstack-tox-lower-constraints/cd37766/ : SUCCESS in 6m 55s\n- openstack-tox-pep8 http://logs.openstack.org/45/654845/1/check/openstack-tox-pep8/21a4fb9/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/45/654845/1/check/openstack-tox-py27/ed90002/ : FAILURE in 5m 15s\n- openstack-tox-py35 http://logs.openstack.org/45/654845/1/check/openstack-tox-py35/24d0ebc/ : FAILURE in 6m 12s\n- openstack-tox-py36 http://logs.openstack.org/45/654845/1/check/openstack-tox-py36/7486f8e/ : FAILURE in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6893832b8280fc7861591d30439e6cc3be4d9ca3","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-04-23 11:11:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"69ef9a4b3ff3e9877c05fa71eb0f8a051b862b71","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-23 18:10:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"150ffc0d90ea2dd24f92a211fa8856029c6b2a08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 18:32:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/45/654845/2/check/test-release-openstack/aa3ddd4/ : SUCCESS in 3m 55s\n- openstack-tox-cover http://logs.openstack.org/45/654845/2/check/openstack-tox-cover/a9236f2/ : FAILURE in 8m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/654845/2/check/openstack-tox-lower-constraints/23ee90e/ : SUCCESS in 7m 54s\n- openstack-tox-pep8 http://logs.openstack.org/45/654845/2/check/openstack-tox-pep8/921a085/ : SUCCESS in 8m 40s\n- openstack-tox-py27 http://logs.openstack.org/45/654845/2/check/openstack-tox-py27/e75e078/ : FAILURE in 6m 26s\n- openstack-tox-py35 http://logs.openstack.org/45/654845/2/check/openstack-tox-py35/3659089/ : FAILURE in 6m 26s\n- openstack-tox-py36 http://logs.openstack.org/45/654845/2/check/openstack-tox-py36/15a246b/ : FAILURE in 6m 45s","accounts_in_message":[],"_revision_number":2},{"id":"0790f9d02f9c34853045176434609f089f95b031","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-24 17:35:54.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"15bfa750fce8e6d324752aa2d52c5e092cf9dd07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 18:20:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/45/654845/2/check/test-release-openstack/c41b6a3/ : SUCCESS in 3m 23s\n- openstack-tox-cover http://logs.openstack.org/45/654845/2/check/openstack-tox-cover/093715a/ : FAILURE in 6m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/654845/2/check/openstack-tox-lower-constraints/18c6f83/ : SUCCESS in 6m 44s\n- openstack-tox-pep8 http://logs.openstack.org/45/654845/2/check/openstack-tox-pep8/098449a/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/45/654845/2/check/openstack-tox-py27/61a2825/ : FAILURE in 7m 09s\n- openstack-tox-py35 http://logs.openstack.org/45/654845/2/check/openstack-tox-py35/22b65d3/ : FAILURE in 6m 23s\n- openstack-tox-py36 http://logs.openstack.org/45/654845/2/check/openstack-tox-py36/e93cddb/ : FAILURE in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"85174aa5131e124a674975333b94ed254c328e4d","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-05-08 12:48:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a81afb68c23766e6b991fcf048464cc37945a48f","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-05-08 12:50:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d68d8c09e72597c3b0254e23bf12ded05c8b0665","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 13:02:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/45/654845/2/check/test-release-openstack/7597028/ : SUCCESS in 3m 06s\n- openstack-tox-cover http://logs.openstack.org/45/654845/2/check/openstack-tox-cover/4747412/ : FAILURE in 7m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/654845/2/check/openstack-tox-lower-constraints/40cfd43/ : SUCCESS in 6m 04s\n- openstack-tox-pep8 http://logs.openstack.org/45/654845/2/check/openstack-tox-pep8/1419689/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/45/654845/2/check/openstack-tox-py27/8afb565/ : FAILURE in 5m 35s\n- openstack-tox-py35 http://logs.openstack.org/45/654845/2/check/openstack-tox-py35/5424d72/ : FAILURE in 5m 17s\n- openstack-tox-py36 http://logs.openstack.org/45/654845/2/check/openstack-tox-py36/d1aa668/ : FAILURE in 5m 47s","accounts_in_message":[],"_revision_number":2},{"id":"eb5106714efd779f9b3c4f05ca30be17c0847d7c","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-05-09 01:57:40.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"250a809591ba681e3dd3809bd2cf201c26c42e59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 02:09:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/45/654845/2/check/test-release-openstack/0718541/ : SUCCESS in 3m 58s\n- openstack-tox-cover http://logs.openstack.org/45/654845/2/check/openstack-tox-cover/938be8a/ : FAILURE in 6m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/654845/2/check/openstack-tox-lower-constraints/02cb202/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 http://logs.openstack.org/45/654845/2/check/openstack-tox-pep8/d996607/ : SUCCESS in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/45/654845/2/check/openstack-tox-py27/1ebc51a/ : FAILURE in 4m 58s\n- openstack-tox-py35 http://logs.openstack.org/45/654845/2/check/openstack-tox-py35/9379209/ : FAILURE in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/45/654845/2/check/openstack-tox-py36/21e58d4/ : FAILURE in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"8ba1a9f6b11cde5f7ca028257bd828fe9fbc792b","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-05-09 11:57:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9466ad947a13d3ee835b288a1869ad0178335baa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 12:11:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/45/654845/2/check/test-release-openstack/9b776d2/ : SUCCESS in 3m 02s\n- openstack-tox-cover http://logs.openstack.org/45/654845/2/check/openstack-tox-cover/5fb5601/ : FAILURE in 6m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/654845/2/check/openstack-tox-lower-constraints/647ba5c/ : SUCCESS in 6m 04s\n- openstack-tox-pep8 http://logs.openstack.org/45/654845/2/check/openstack-tox-pep8/056f6b5/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/45/654845/2/check/openstack-tox-py27/c56135d/ : FAILURE in 5m 06s\n- openstack-tox-py35 http://logs.openstack.org/45/654845/2/check/openstack-tox-py35/6a72748/ : FAILURE in 7m 32s\n- openstack-tox-py36 http://logs.openstack.org/45/654845/2/check/openstack-tox-py36/e63cae7/ : FAILURE in 5m 37s","accounts_in_message":[],"_revision_number":2},{"id":"d9b6dc5ef8b1fbd68488cb60f121ef40e89ee12b","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-05-09 12:41:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4e738085a9ef7ecb09738429664342309034c9e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 12:53:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/45/654845/2/check/test-release-openstack/8a72cc7/ : SUCCESS in 3m 19s\n- openstack-tox-cover http://logs.openstack.org/45/654845/2/check/openstack-tox-cover/4d3fbe1/ : FAILURE in 6m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/654845/2/check/openstack-tox-lower-constraints/333376d/ : SUCCESS in 5m 49s\n- openstack-tox-pep8 http://logs.openstack.org/45/654845/2/check/openstack-tox-pep8/adc66e5/ : SUCCESS in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/45/654845/2/check/openstack-tox-py27/b4ccacc/ : FAILURE in 5m 00s\n- openstack-tox-py35 http://logs.openstack.org/45/654845/2/check/openstack-tox-py35/89c3d82/ : FAILURE in 5m 22s\n- openstack-tox-py36 http://logs.openstack.org/45/654845/2/check/openstack-tox-py36/47770c1/ : FAILURE in 5m 36s","accounts_in_message":[],"_revision_number":2},{"id":"79539b14e1bc23df9be1933c7c405f4a2569e4a6","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-05-10 08:30:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3b9bf23446085b9213724a3c8531cf66ca35056a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 08:42:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/45/654845/2/check/test-release-openstack/d4c1d46/ : SUCCESS in 3m 14s\n- openstack-tox-cover http://logs.openstack.org/45/654845/2/check/openstack-tox-cover/ecf8e2d/ : FAILURE in 6m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/654845/2/check/openstack-tox-lower-constraints/7cd6a42/ : SUCCESS in 6m 28s\n- openstack-tox-pep8 http://logs.openstack.org/45/654845/2/check/openstack-tox-pep8/a801596/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/45/654845/2/check/openstack-tox-py27/65239d3/ : FAILURE in 5m 40s\n- openstack-tox-py35 http://logs.openstack.org/45/654845/2/check/openstack-tox-py35/e5ef386/ : FAILURE in 5m 39s\n- openstack-tox-py36 http://logs.openstack.org/45/654845/2/check/openstack-tox-py36/344878d/ : FAILURE in 5m 36s","accounts_in_message":[],"_revision_number":2},{"id":"9fa3f34ee80c557e1607f4a0338876cfc3aef46d","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-11 08:06:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"88dc4772277e608c13549f9006441dc752545695","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 08:18:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/45/654845/2/check/test-release-openstack/157e09a/ : SUCCESS in 3m 01s\n- openstack-tox-cover http://logs.openstack.org/45/654845/2/check/openstack-tox-cover/63b5239/ : FAILURE in 6m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/654845/2/check/openstack-tox-lower-constraints/8a160ee/ : SUCCESS in 6m 03s\n- openstack-tox-pep8 http://logs.openstack.org/45/654845/2/check/openstack-tox-pep8/aa8b008/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/45/654845/2/check/openstack-tox-py27/75fcb7f/ : FAILURE in 5m 03s\n- openstack-tox-py35 http://logs.openstack.org/45/654845/2/check/openstack-tox-py35/e646d7d/ : FAILURE in 5m 11s\n- openstack-tox-py36 http://logs.openstack.org/45/654845/2/check/openstack-tox-py36/87db420/ : FAILURE in 5m 11s","accounts_in_message":[],"_revision_number":2},{"id":"4e9022ec477e2a31d5bac0fc8d4e46efbb209966","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-05-13 13:16:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a6115f9bbdce0ed9257a07e251672ac3708ed35c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 13:34:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/45/654845/2/check/test-release-openstack/4316971/ : SUCCESS in 8m 55s\n- openstack-tox-cover http://logs.openstack.org/45/654845/2/check/openstack-tox-cover/6c9218e/ : FAILURE in 17m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/654845/2/check/openstack-tox-lower-constraints/f7af0be/ : SUCCESS in 5m 24s\n- openstack-tox-pep8 http://logs.openstack.org/45/654845/2/check/openstack-tox-pep8/a422660/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/45/654845/2/check/openstack-tox-py27/bd542de/ : FAILURE in 5m 00s\n- openstack-tox-py35 http://logs.openstack.org/45/654845/2/check/openstack-tox-py35/eaec39c/ : FAILURE in 4m 56s\n- openstack-tox-py36 http://logs.openstack.org/45/654845/2/check/openstack-tox-py36/7630c32/ : FAILURE in 5m 39s","accounts_in_message":[],"_revision_number":2},{"id":"48949b8f680dd3ad738633487bbec2e127656c2e","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-05-13 13:46:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fcb6d29e8589fc2e2804b6addc4cb5cb6f9b6978","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-14 02:57:02.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2d8c215a02a023ce8eedfda9355d18c36fb982cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 03:04:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/45/654845/2/check/test-release-openstack/8e02682/ : SUCCESS in 4m 36s\n- openstack-tox-cover http://logs.openstack.org/45/654845/2/check/openstack-tox-cover/63494c9/ : FAILURE in 5m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/654845/2/check/openstack-tox-lower-constraints/841dc43/ : SUCCESS in 7m 03s\n- openstack-tox-pep8 http://logs.openstack.org/45/654845/2/check/openstack-tox-pep8/1e4aa57/ : SUCCESS in 6m 18s\n- openstack-tox-py27 http://logs.openstack.org/45/654845/2/check/openstack-tox-py27/e7f6f04/ : FAILURE in 5m 37s\n- openstack-tox-py35 http://logs.openstack.org/45/654845/2/check/openstack-tox-py35/48131ac/ : FAILURE in 6m 07s\n- openstack-tox-py36 http://logs.openstack.org/45/654845/2/check/openstack-tox-py36/28f8c47/ : FAILURE in 6m 31s","accounts_in_message":[],"_revision_number":2},{"id":"067bcc7d8a5e6e71658b97c56e627aa496756027","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-05-14 08:07:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2e64260455783fa3c4c6c22b38e8e3cd91ff985f","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-19 10:09:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3efcfea07ed14634001ee20425624aa5cebdc682","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 10:40:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/45/654845/2/check/test-release-openstack/ca7b767/ : SUCCESS in 3m 32s\n- openstack-tox-cover http://logs.openstack.org/45/654845/2/check/openstack-tox-cover/498d73e/ : FAILURE in 6m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/654845/2/check/openstack-tox-lower-constraints/ea0d40d/ : SUCCESS in 6m 29s\n- openstack-tox-pep8 http://logs.openstack.org/45/654845/2/check/openstack-tox-pep8/186288b/ : SUCCESS in 7m 04s\n- openstack-tox-py27 http://logs.openstack.org/45/654845/2/check/openstack-tox-py27/6540649/ : FAILURE in 5m 00s\n- openstack-tox-py35 http://logs.openstack.org/45/654845/2/check/openstack-tox-py35/ffc83fc/ : FAILURE in 5m 25s\n- openstack-tox-py36 http://logs.openstack.org/45/654845/2/check/openstack-tox-py36/70e2f42/ : FAILURE in 6m 14s","accounts_in_message":[],"_revision_number":2},{"id":"124e1bf13c8cf6d50f1fcc5e5ce9e91004d662df","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-05-20 14:17:00.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"7e2d7cd6ac9c51901dc226756e809770b4609fcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 14:29:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/45/654845/3/check/test-release-openstack/20ae4af/ : SUCCESS in 2m 58s\n- openstack-tox-cover http://logs.openstack.org/45/654845/3/check/openstack-tox-cover/da85ef7/cover/ : SUCCESS in 6m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/654845/3/check/openstack-tox-lower-constraints/04c010b/ : SUCCESS in 6m 24s\n- openstack-tox-pep8 http://logs.openstack.org/45/654845/3/check/openstack-tox-pep8/ab628ef/ : SUCCESS in 5m 53s\n- openstack-tox-py27 http://logs.openstack.org/45/654845/3/check/openstack-tox-py27/49f2816/ : SUCCESS in 5m 59s\n- openstack-tox-py35 http://logs.openstack.org/45/654845/3/check/openstack-tox-py35/977bd30/ : SUCCESS in 5m 28s\n- openstack-tox-py36 http://logs.openstack.org/45/654845/3/check/openstack-tox-py36/dd84df3/ : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":3},{"id":"87e2fbb36fab6d548f58408aa2b195699dba73ff","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-05-20 15:04:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5e709ba1fecaffdd645c2a3a80a419ff500ef494","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 15:05:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"394249c5d472f90cded819da69e27f073cf5511e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 15:17:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/45/654845/3/gate/test-release-openstack/46d2a2c/ : SUCCESS in 3m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/654845/3/gate/openstack-tox-lower-constraints/e45998f/ : SUCCESS in 5m 59s\n- openstack-tox-pep8 http://logs.openstack.org/45/654845/3/gate/openstack-tox-pep8/3d406e9/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/45/654845/3/gate/openstack-tox-py27/eba0d34/ : SUCCESS in 8m 32s\n- openstack-tox-py35 http://logs.openstack.org/45/654845/3/gate/openstack-tox-py35/1fbd92d/ : SUCCESS in 5m 38s\n- openstack-tox-py36 http://logs.openstack.org/45/654845/3/gate/openstack-tox-py36/193fcbc/ : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":3},{"id":"49fe554a5f2bc41724cdffe92b9932dc6aab7edc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 15:17:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9d1f6ea4f0dd17ee3b81d7178fc131057d27c96f","revisions":{"03438b0ef8925e61cd6686b6a9fd0f933ac3b4b5":{"kind":"REWORK","_number":1,"created":"2019-04-23 08:41:30.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/45/654845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/45/654845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/45/654845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/45/654845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/45/654845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/45/654845/1"}}},"commit":{"parents":[{"commit":"b7bedc0e810121b3382c776ac15eade7e031b0f6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/b7bedc0e810121b3382c776ac15eade7e031b0f6"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:41:16.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:41:27.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I08632e63eef7cb81f0b3828db69feb9750d65cb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/03438b0ef8925e61cd6686b6a9fd0f933ac3b4b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/03438b0ef8925e61cd6686b6a9fd0f933ac3b4b5"}]},"branch":"refs/heads/master"},"ca2aebc7876741122dada20a2d4917eaad5012dc":{"kind":"REWORK","_number":2,"created":"2019-04-23 18:10:03.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/45/654845/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/45/654845/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/45/654845/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/45/654845/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/45/654845/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/45/654845/2"}}},"commit":{"parents":[{"commit":"b7bedc0e810121b3382c776ac15eade7e031b0f6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/b7bedc0e810121b3382c776ac15eade7e031b0f6"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:41:16.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 18:09:57.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I08632e63eef7cb81f0b3828db69feb9750d65cb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/ca2aebc7876741122dada20a2d4917eaad5012dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/ca2aebc7876741122dada20a2d4917eaad5012dc"}]},"branch":"refs/heads/master"},"9d1f6ea4f0dd17ee3b81d7178fc131057d27c96f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-20 14:17:00.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/45/654845/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/45/654845/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/45/654845/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/45/654845/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/45/654845/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/45/654845/3"}}},"commit":{"parents":[{"commit":"c85ecafe59a580042ac3860c67b3f1cc44dd5830","subject":"Merge \"Fix failing unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/c85ecafe59a580042ac3860c67b3f1cc44dd5830"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:41:16.000000000","tz":480},"committer":{"name":"Witold Bedyk","email":"witold.bedyk@suse.com","date":"2019-05-20 14:17:00.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I08632e63eef7cb81f0b3828db69feb9750d65cb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/9d1f6ea4f0dd17ee3b81d7178fc131057d27c96f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/9d1f6ea4f0dd17ee3b81d7178fc131057d27c96f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
