)]}'
{"id":"openstack%2Fmonasca-ui~700637","triplet_id":"openstack%2Fmonasca-ui~master~I84d36a14b779a510bea5a24d991192016ae73273","project":"openstack/monasca-ui","branch":"master","topic":"babel-config","hashtags":[],"change_id":"I84d36a14b779a510bea5a24d991192016ae73273","subject":"translation: drop babel extractor definitions","status":"MERGED","created":"2019-12-26 17:19:28.000000000","updated":"2020-01-08 08:39:33.000000000","submitted":"2020-01-08 08:39:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700637-1578472773549-d94f54cc","meta_rev_id":"5d1b286c3eddfffcc057493a3659f065ae18bd13","_number":700637,"virtual_id_number":700637,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-08 08:39:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2019-12-27 20:27:27.000000000","_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-01-08 08:29:21.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2020-01-08 08:29:21.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-27 20:27:27.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-01-08 08:39:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb216df5000076a27b070ed0bdcf6ffab1cc8318","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-26 17:19:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c5993c2f4de32a11abb9a30e24da8f6284b9ea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 17:27:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/5f43f296aa9a419bad4fd94513635dbc : SUCCESS in 4m 50s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/391af4326dbb44f9b73e3432456c8d9c : SUCCESS in 4m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e1b4e6c8953408d9e6725fc443cbd92 : SUCCESS in 7m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5ffca3949e14719b4ac8a5c4eb4a5a9 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/987dbe4138d94b708c2a0a9e1046edb3 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94ba28a0169e448a852b97e7e56e7bb7 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cfeab3377f9747eb8879577adef70cce : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5d928a596701da579c776716e3710bee5825b39b","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-27 20:27:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5a761f8746a6b8859577c53ec7c2dcf4f856fdec","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-08 08:29:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nPushing it into the gate to make the translation job again.\nI am self-approving as it is a maintenance change to recover the infra job.","accounts_in_message":[],"_revision_number":1},{"id":"dc5383a732956fc46dedd3dcf72bf8ce2e33df36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 08:29:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ad80de49cfdc998b472f5575d3be033062e0c02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 08:39:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0a9d1f04a02745a5645c271bb782a48dc26afb60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 08:39:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/3da5f52882df4289bf9c6415b60a0d8a : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1704557a205e4557b27b7e3681180a29 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a77dd7294589433990f0d9e6ca19c3c6 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ca8ca0e5c674546816bc853a1bc630e : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/482c2c9b53374012996ca350d1843fa5 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d4bd0bd25758e84637415b38140c37c794d7356","revisions":{"0d4bd0bd25758e84637415b38140c37c794d7356":{"kind":"REWORK","_number":1,"created":"2019-12-26 17:19:28.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/37/700637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/37/700637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/37/700637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/37/700637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/37/700637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/37/700637/1"}}},"commit":{"parents":[{"commit":"4e82a8103e8cc1eac93c17cfeb1b22de6affcf87","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/4e82a8103e8cc1eac93c17cfeb1b22de6affcf87"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-26 17:18:23.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-26 17:18:23.000000000","tz":540},"subject":"translation: drop babel extractor definitions","message":"translation: drop babel extractor definitions\n\nbabel extractors are now registered via python entry points,\nso there is no need to declare babel extractors in babel configs.\n\nThis change is important to make translation work in Django 2.2.\ndjango-babel does not work with Django 2.2 and looks unmaintained\nfor over two years. The horizon team is thinking to switch the extractor\nto enmerkar (a fork of django-babel) to make extraction of translation\nstring work again near future. It is important to drop the extractor\ndefinition to make the transition smooth.\n\nChange-Id: I84d36a14b779a510bea5a24d991192016ae73273\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/0d4bd0bd25758e84637415b38140c37c794d7356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/0d4bd0bd25758e84637415b38140c37c794d7356"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
