)]}'
{"id":"openstack%2Fmonasca-ui~700929","triplet_id":"openstack%2Fmonasca-ui~master~I3cd4ad5c3fe590cb1522b2be1b2059d7c70b45cc","project":"openstack/monasca-ui","branch":"master","topic":"drop-django111-support","hashtags":[],"change_id":"I3cd4ad5c3fe590cb1522b2be1b2059d7c70b45cc","subject":"Drop Django 1.11 support","status":"MERGED","created":"2020-01-02 19:40:52.000000000","updated":"2020-01-17 15:30:05.000000000","submitted":"2020-01-17 15:30:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700929-1579275005667-641ca554","meta_rev_id":"41e64efc97059148d273cf7dd4b90e80181c0784","_number":700929,"virtual_id_number":700929,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-17 15:30:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"recommended":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"all":[{"value":1,"date":"2020-01-03 09:23:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-15 08:41:55.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-17 15:19:54.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-03 09:23:34.000000000","updated_by":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"reviewer":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"state":"REVIEWER"},{"updated":"2020-01-17 15:19:54.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2020-01-17 15:30:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"594c87cab9b3f1e8e283c58fb99ad499006384b2","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-02 19:40:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34181d2f48e8265e62ecdd47a869b767de0d4071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 19:49:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ba11a2a6146949449442e40263214283 : SUCCESS in 2m 23s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/74fcc7496fb942588e43f8a27eae7313 : SUCCESS in 4m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bffd90d9744549b79a4946974a3b5e12 : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4298324e6864901bf0eeb1fda18e11a : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40cd6225031344e0bba3f0f32b985404 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1563502ef26244c8a86381c20844ec2f : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78f6f2b0a7e143bf8a03b86ed0a67c1a : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"20ca9a60a4adaaca2e40350a3f27c9ab19cf60d8","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-02 20:56:26.000000000","message":"Topic changed from django22 to drop-django111-support","accounts_in_message":[],"_revision_number":1},{"id":"9c00560111124cee9a32670be6bf534693eadbb3","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2020-01-03 09:23:34.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"507b2880f200e4e17afd716f7c1a400309905894","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-01-15 08:41:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3630db65e62111871b1e7913a46eb68cfba36ac0","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-01-17 15:19:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a6d6fa4683b26ca3e4a378c581b41adc0e2d5df2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 15:20:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d3b4259dab2892c55ac3ca3a317197bbe9d2b7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 15:30:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"41e64efc97059148d273cf7dd4b90e80181c0784","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 15:30:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/091be6bdaae44b999c4b9cc138f2d465 : SUCCESS in 4m 23s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/1c76571a978d4537a80934cdb216385f : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db36fec063c8492cb2494efe5778bde5 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aabb25c4a58e4b48a280ffebcccc5fa6 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e34011850bea4494bffdc488f68a42db : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e813bd5ec5b34dd49387998e464a78a4 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8d3417d0b6f081f28f87e0049daab01b20d161a","revisions":{"f8d3417d0b6f081f28f87e0049daab01b20d161a":{"kind":"REWORK","_number":1,"created":"2020-01-02 19:40:52.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/29/700929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/29/700929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/29/700929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/29/700929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/29/700929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/29/700929/1"}}},"commit":{"parents":[{"commit":"4e82a8103e8cc1eac93c17cfeb1b22de6affcf87","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/4e82a8103e8cc1eac93c17cfeb1b22de6affcf87"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-02 19:40:26.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-02 19:40:26.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2. requirements.txt and\nlower-constraints.txt are updated accordingly. for more info. please\nrefer [1].\nDepends-On: https://review.opendev.org/#/c/700733/\n[1] https://etherpad.openstack.org/p/Enable_Django22_support_in_Horizon_Plugin\n\nChange-Id: I3cd4ad5c3fe590cb1522b2be1b2059d7c70b45cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/f8d3417d0b6f081f28f87e0049daab01b20d161a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/f8d3417d0b6f081f28f87e0049daab01b20d161a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
