)]}'
{"id":"openstack%2Fmonasca-ui~722510","triplet_id":"openstack%2Fmonasca-ui~master~I03f0d04801023a88d48ee9692164b7359043adc6","project":"openstack/monasca-ui","branch":"master","topic":"add-victoria-python-jobtemplates","hashtags":[],"change_id":"I03f0d04801023a88d48ee9692164b7359043adc6","subject":"Add Python3 victoria unit tests","status":"MERGED","created":"2020-04-23 21:30:18.000000000","updated":"2020-05-28 13:02:49.000000000","submitted":"2020-05-28 13:02:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722510-1590670969343-41f9ad85","meta_rev_id":"7521f8d829f56bb1d9e510e34262cc7155305755","_number":722510,"virtual_id_number":722510,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 13:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2020-05-08 12:57:55.000000000","_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-28 12:52:54.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-08 12:57:55.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-05-28 12:52:54.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2020-05-28 13:02:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f29abb81b137675fece837a51546971fe6acb34b","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-23 21:30:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"905ea154062239b4086328c519beaa895dd5c4a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 21:47:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b857cfa95464914a8a1af31ba32dfc7 : SUCCESS in 6m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2be5d445fa84eba9c958e807cdabf70 : SUCCESS in 9m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ece5a8f8142473cb4e9b741f3ef79e3 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bf1d7ec5a2e49cd81d05ed8b2091c6d : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/340231cb02fa4005ba8c22d2caa3b4ab : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3674629c4b0400ab1464d3a05f640a0 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":1},{"id":"34ae8cfc5689e803c5502b28290af772b6135c0c","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-05-08 12:57:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"539b2dfd9a19741a5788b14358c52dffa6392c8a","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-05-28 12:52:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"440f6239b6f28e1a929ad18c29b1da684ad15cd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 12:53:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1dbce80ec67dc09fa1b851db3cad14f03cb3aae1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:02:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07219c2f4e32490e9c8670e853576731 : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73766de5201746daa9461304f2af8b5f : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bebd97d808c74ace91e0263d228c5e77 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3f4fab62d92447da7dd6f465da02d96 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c2661aadf6c0af5654dae205b9226b55dc285790","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:02:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68c9fb4c268e35a7de90f2b31122eef06537a9a0","revisions":{"68c9fb4c268e35a7de90f2b31122eef06537a9a0":{"kind":"REWORK","_number":1,"created":"2020-04-23 21:30:18.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/10/722510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/10/722510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/10/722510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/10/722510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/10/722510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/10/722510/1"}}},"commit":{"parents":[{"commit":"135d98dfa317413d1e27c887241fd881c577eb3b","subject":"Merge \"Fix pyScss version in lower-constraints.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/135d98dfa317413d1e27c887241fd881c577eb3b"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 21:30:16.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 21:30:16.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I03f0d04801023a88d48ee9692164b7359043adc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/68c9fb4c268e35a7de90f2b31122eef06537a9a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/68c9fb4c268e35a7de90f2b31122eef06537a9a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
