)]}'
{"id":"openstack%2Fmonasca-ui~748489","triplet_id":"openstack%2Fmonasca-ui~master~I1ddea97cdcaf4142d9dabcf4a629e805a9c1e152","project":"openstack/monasca-ui","branch":"master","topic":"bp/refactor-error-messages","hashtags":[],"change_id":"I1ddea97cdcaf4142d9dabcf4a629e805a9c1e152","subject":"Cleanup for Refactor-error-messages","status":"MERGED","created":"2020-08-27 17:15:45.000000000","updated":"2020-10-14 13:16:22.000000000","submitted":"2020-10-14 13:16:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748489-1602681382875-eaa21e4c","meta_rev_id":"7cc907d5ba41dcf6e31bb5b31269bd083683cb31","_number":748489,"virtual_id_number":748489,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-14 13:16:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-09-16 15:02:19.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-14 13:05:13.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-15 09:34:56.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-09-16 15:02:19.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-10-14 13:05:13.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2020-10-14 13:16:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c05fe7cb32cced3413b6b017c921da6a42662ec3","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-27 17:15:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24e96662ca27ca6bc5e763af69b86f4362ce374d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-27 17:16:37.000000000","message":"Topic set to bp/refactor-error-messages","accounts_in_message":[],"_revision_number":1},{"id":"f95bcb5f056fe327c9052554fc30a75bf29a6263","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 17:34:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d97d97098c9b4262bff5fc9cd1a5f5b4 : SUCCESS in 9m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7dfe9716cad1472b8c97cfaff6373259 : SUCCESS in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/086ab7f60258461a8bfeb69fed8da3c9 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ab05502319f47e0b28e3ba79c724b1d : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bd2c9c5a4fc4538ad85496575e47c6c : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"43bd4456dba972a4a55f9a42f2b8087f37981286","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-08 11:02:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c9407fd275d2141defa51f619e24d1592d1b2288","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-12 01:33:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"626036d101d0d50d9f798ed4db1ff247b56656b5","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-14 05:14:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2ad7e094222a28d40fbd998817f43fc97379a8d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-14 05:14:55.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a9c8b392d66a903177b06906903c5c67a688e7e0","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-14 05:19:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b6dd2814499928a156c7fa961ae0d187eafb1662","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-14 05:21:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"33b04dea25ba9afd67f435bd5cc3ff0f717eef10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 05:33:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef13f3abc0824af78c75d442ab3ce536 : SUCCESS in 3m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc587bac6c724f7e9d09df12ac95b13a : SUCCESS in 6m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/921524151d504d63b0202bf4929e8dca : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbefd3c553ff40e4a85eb6ecea00f093 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f70238ca3aaa4949b8ba0a8ac13efd6a : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":4},{"id":"235b63c69b9c0532a9323c1cf80b200087b0a59b","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-14 08:35:38.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a42e3c7a7eeb20303a5a988023971c8d5f2a00e0","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-15 09:34:56.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6adf15fb2500bdaa766b02c1bb0133d12a105f4d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-16 08:20:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6ce4763a14ae844a04c210cd696a4bc053b83f10","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-16 08:20:11.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"eba8a8922ac867a938fbb1df3f9844a19cb89c26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 08:37:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b5705a4ede34c25863fa1e3edb37df5 : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c77f7cae9d0f4c68acabf8435667a595 : SUCCESS in 11m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eddb2642f12d4af390a5b4c5be803975 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9137cad95d974e30b6a05ba754b3457d : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3617c266dbd442082a877ef3d66b341 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":5},{"id":"8d431a5c90911ccbbb60b1726ec60394da4bd67b","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-16 15:02:19.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b33659b1e880f0e560d5dd2b14630383a55157d1","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-22 05:02:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"227cb8d2f6d05d505927b0596873c555dfd9545b","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-10-14 13:05:13.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e05c51e1d9b7ba60f91bdf5992ca8d1e6971ce80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 13:05:25.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1fba1b422eee65884e58664ebc0320ae0fec0244","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 13:16:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"7cc907d5ba41dcf6e31bb5b31269bd083683cb31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 13:16:22.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b504e0ce2a8640ca80af9cb11c1c06a4 : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d63f3ad8b35547ae903cfc8e15ff0773 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de5d179b0554441c9119050ff4464f26 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45590469e8974702bfb54e784673599d : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b4991fcf250eb4c033b19a3b29a175995a97fcfa","revisions":{"d75b2f223f2271074dc2bfb2ae438929111ca534":{"kind":"REWORK","_number":1,"created":"2020-08-27 17:15:45.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/89/748489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/89/748489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/1"}}},"commit":{"parents":[{"commit":"2bfbee4bb9576fff6c0eeaf4a308ba030a1505a6","subject":"Alarm definition breakes Alarms view","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/2bfbee4bb9576fff6c0eeaf4a308ba030a1505a6"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-08-27 17:12:25.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-08-27 17:12:25.000000000","tz":0},"subject":"Cleanup for Refactor-error-messages","message":"Cleanup for Refactor-error-messages\n\nThis patch is a clean-up patch for refactor-error-messages bp\nwhich remove the exception message from base message otherwise\nthe same exception message display twice like\nthis https://ibb.co/XyFWMdz .\n\nDepends-On: https://review.opendev.org/#/c/708069/\nChange-Id: I1ddea97cdcaf4142d9dabcf4a629e805a9c1e152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/d75b2f223f2271074dc2bfb2ae438929111ca534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/d75b2f223f2271074dc2bfb2ae438929111ca534"}]},"branch":"refs/heads/master"},"2beb027a52ab81f5c2e43f64f8f58fd79489c643":{"kind":"REWORK","_number":2,"created":"2020-09-14 05:14:45.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/89/748489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/89/748489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/2"}}},"commit":{"parents":[{"commit":"2bfbee4bb9576fff6c0eeaf4a308ba030a1505a6","subject":"Alarm definition breakes Alarms view","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/2bfbee4bb9576fff6c0eeaf4a308ba030a1505a6"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-08-27 17:12:25.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-09-14 05:14:28.000000000","tz":0},"subject":"Cleanup for Refactor-error-messages","message":"Cleanup for Refactor-error-messages\n\nThis patch is a clean-up patch for refactor-error-messages bp\nwhich remove the exception message from base message otherwise\nthe same exception message display twice like\nthis https://ibb.co/XyFWMdz .\n\nDepends-On: https://review.opendev.org/#/c/708069/\nChange-Id: I1ddea97cdcaf4142d9dabcf4a629e805a9c1e152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/2beb027a52ab81f5c2e43f64f8f58fd79489c643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/2beb027a52ab81f5c2e43f64f8f58fd79489c643"}]},"branch":"refs/heads/master"},"3e8d6b1c0ebaf67f3c4dee816dcbff06630f1f21":{"kind":"REWORK","_number":3,"created":"2020-09-14 05:19:24.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/89/748489/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/89/748489/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/3"}}},"commit":{"parents":[{"commit":"2bfbee4bb9576fff6c0eeaf4a308ba030a1505a6","subject":"Alarm definition breakes Alarms view","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/2bfbee4bb9576fff6c0eeaf4a308ba030a1505a6"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-08-27 17:12:25.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-09-14 05:19:11.000000000","tz":0},"subject":"Cleanup for Refactor-error-messages","message":"Cleanup for Refactor-error-messages\n\nThis patch is a clean-up patch for refactor-error-messages bp\nwhich remove the exception message from base message otherwise\nthe same exception message display twice like\nthis https://ibb.co/XyFWMdz .\n\nDepends-On: https://review.opendev.org/#/c/708069/\nChange-Id: I1ddea97cdcaf4142d9dabcf4a629e805a9c1e152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/3e8d6b1c0ebaf67f3c4dee816dcbff06630f1f21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/3e8d6b1c0ebaf67f3c4dee816dcbff06630f1f21"}]},"branch":"refs/heads/master"},"fb7fa4c5fbcd61ce2aef6f4ab2c5225781c81413":{"kind":"REWORK","_number":4,"created":"2020-09-14 05:21:46.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/89/748489/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/89/748489/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/4"}}},"commit":{"parents":[{"commit":"2bfbee4bb9576fff6c0eeaf4a308ba030a1505a6","subject":"Alarm definition breakes Alarms view","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/2bfbee4bb9576fff6c0eeaf4a308ba030a1505a6"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-08-27 17:12:25.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-09-14 05:21:30.000000000","tz":0},"subject":"Cleanup for Refactor-error-messages","message":"Cleanup for Refactor-error-messages\n\nThis patch is a clean-up patch for refactor-error-messages bp\nwhich remove the exception message from base message otherwise\nthe same exception message display twice like\nthis https://ibb.co/XyFWMdz .\n\nDepends-On: https://review.opendev.org/#/c/708069/\nChange-Id: I1ddea97cdcaf4142d9dabcf4a629e805a9c1e152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/fb7fa4c5fbcd61ce2aef6f4ab2c5225781c81413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/fb7fa4c5fbcd61ce2aef6f4ab2c5225781c81413"}]},"branch":"refs/heads/master"},"b4991fcf250eb4c033b19a3b29a175995a97fcfa":{"kind":"REWORK","_number":5,"created":"2020-09-16 08:20:09.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/89/748489/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/89/748489/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/89/748489/5"}}},"commit":{"parents":[{"commit":"2bfbee4bb9576fff6c0eeaf4a308ba030a1505a6","subject":"Alarm definition breakes Alarms view","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/2bfbee4bb9576fff6c0eeaf4a308ba030a1505a6"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-08-27 17:12:25.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-09-16 08:19:57.000000000","tz":0},"subject":"Cleanup for Refactor-error-messages","message":"Cleanup for Refactor-error-messages\n\nThis patch is a clean-up patch for refactor-error-messages bp\nwhich remove the exception message from base message otherwise\nthe same exception message display twice like\nthis https://ibb.co/XyFWMdz .\n\nDepends-On: https://review.opendev.org/#/c/708069/\nChange-Id: I1ddea97cdcaf4142d9dabcf4a629e805a9c1e152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/b4991fcf250eb4c033b19a3b29a175995a97fcfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/b4991fcf250eb4c033b19a3b29a175995a97fcfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
