)]}'
{"id":"openstack%2Fmonasca-ui~754761","triplet_id":"openstack%2Fmonasca-ui~master~Id81a44c484ce86436c6f2f3278c96b3ffc85abb1","project":"openstack/monasca-ui","branch":"master","topic":"add-wallaby-python-jobtemplates","hashtags":[],"change_id":"Id81a44c484ce86436c6f2f3278c96b3ffc85abb1","subject":"Add Python3 wallaby unit tests","status":"MERGED","created":"2020-09-28 15:47:53.000000000","updated":"2020-10-13 16:39:17.000000000","submitted":"2020-10-13 16:39:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754761-1602607157996-332a2ac9","meta_rev_id":"5cdc9f8aeb850fd575e6ff341671e919b8f20f31","_number":754761,"virtual_id_number":754761,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-13 16:39:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2020-10-05 09:42:34.000000000","_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-13 16:20:16.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-13 16:20:16.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 09:42:34.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-10-13 16:20:16.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2020-10-13 16:39:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b65dc66bd06d290ccc6f21fad0e162c34dd3fcc","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-28 15:47:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"008ae2b82cd23c1a71eb36e97923dc80a2fda008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 16:29:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48c38e4783c0496c94bd2b777f086be4 : SUCCESS in 4m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06945000b43944519f0ba2296319789a : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7643a3a203094dc38cf3908d30b439fd : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cd8f80cfd164ea0876573a97e2c8317 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e146f5f4b20445ae976ccb5dce1c021c : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2a4109cce6823bf9f0535ed31f207df66fcf0655","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-10-05 09:42:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f6d8e7b7356956e5eca41d267e1da4d17fcb8eb5","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-10-13 16:20:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f55fb83a3bdacef95acd2ae20d136b317f511d0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 16:22:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8913e000d57fac70fc3cd78fed36d4f2b32f22dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 16:39:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"032223eeced555d12a44df419cb88c0c8d825154","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 16:39:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9d287983fde486c97a53f62c647d83d : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08c7a6e0440f4ad5a5cceb7def5cdf1f : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e96d2be1d1c94602b8049b03f6dd4b28 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1d8b50ac7a8477cb51c1a95868c2f59 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1da1b7f9e3f2e70d70dd4f985bbb28fb8854bb7e","revisions":{"1da1b7f9e3f2e70d70dd4f985bbb28fb8854bb7e":{"kind":"REWORK","_number":1,"created":"2020-09-28 15:47:53.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/61/754761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/61/754761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/61/754761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/61/754761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/61/754761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/61/754761/1"}}},"commit":{"parents":[{"commit":"0ae1ab35e4877d0e54a1acb2a855bfb8a00dfd9c","subject":"Merge \"[goal] Migrate testing to ubuntu focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/0ae1ab35e4877d0e54a1acb2a855bfb8a00dfd9c"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-28 15:47:52.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-28 15:47:52.000000000","tz":0},"subject":"Add Python3 wallaby unit tests","message":"Add Python3 wallaby unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for wallaby.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Id81a44c484ce86436c6f2f3278c96b3ffc85abb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/1da1b7f9e3f2e70d70dd4f985bbb28fb8854bb7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/1da1b7f9e3f2e70d70dd4f985bbb28fb8854bb7e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
