)]}'
{"id":"openstack%2Fmonasca-ui~761044","triplet_id":"openstack%2Fmonasca-ui~master~I8d2357e51d570e9e0513326248878fd0133144bb","project":"openstack/monasca-ui","branch":"master","topic":"remove-six","hashtags":[],"change_id":"I8d2357e51d570e9e0513326248878fd0133144bb","subject":"Remove six","status":"MERGED","created":"2020-11-03 03:49:55.000000000","updated":"2020-11-03 10:55:05.000000000","submitted":"2020-11-03 10:55:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761044-1604400905420-b6ee05d8","meta_rev_id":"19b42cb0b09bd4617655eaa5f7501b85a76066c6","_number":761044,"virtual_id_number":761044,"owner":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-03 10:55:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-11-03 10:41:22.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"value":2,"date":"2020-11-03 10:25:44.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-11-03 10:41:22.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-03 10:25:44.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2020-11-03 10:25:56.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2020-11-03 10:26:06.000000000","updated_by":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"reviewer":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"state":"REVIEWER"},{"updated":"2020-11-03 10:41:22.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-11-03 10:55:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8693f21a06aa37007f829eb6c93f14b8de1cb1a1","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-11-03 03:49:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e72ae91cce7aa4569f34f84ae8cac23e9465112e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 04:06:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cda19f512f874353a5c1a0518e669f16 : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70918f1dafc8400cafd22e348adccb96 : SUCCESS in 5m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8377e658e8b743c599b684916dd88d2e : SUCCESS in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a21d0a92167473ca911627e25d4086f : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9800bb512c546728f89af54658d9ef9 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a716505dddbd4ee49da9fd92e9a64f36 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/093a6064fdf24dc2bbfeb775dfe0d7b7 : SUCCESS in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e6f41341ab10d235d5d53d75a08bbf0ca16fe22","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-11-03 08:21:55.000000000","message":"Topic set to remove-six","accounts_in_message":[],"_revision_number":1},{"id":"07237f98bdde0474f418f8b9ed3cc9213b23055b","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-11-03 10:22:56.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e2e58960f2d8537ef9b8434af3de244dd561ddf8","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-11-03 10:25:44.000000000","message":"Patch Set 2: Code-Review+2\n\nHi Zihao, thanks for the change. I added the Story/Task to the message.","accounts_in_message":[],"_revision_number":2},{"id":"dcf909d1a8d760dbed4de074f9eba0e68f204cff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 10:37:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d8383b02e5564b4c97fa73d182e00d36 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/807cbf8d42de4048998becd08d77726d : SUCCESS in 6m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a526a0e2ac943929452a76f2dd2d1df : SUCCESS in 7m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c63340ea95840f1a7a5fedab08c498e : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/523e0fcdc1a9451fb2dd2ae29754762b : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c223486da81040f7a78eb8c9800a3df0 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68f6e3bd278641b8977bf83fc27fa5e3 : SUCCESS in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b0476e49617dc69f6fb902b6e7812351f3bc5696","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-11-03 10:41:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"45a2eb28a5dff2d5d610be08e1c20a666c98c058","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 10:41:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6d84339d6598a91d6e2b46ebc2f813675497500d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 10:55:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/dd843ded0c7c46b3bb216147f6244c2a : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/002a2c03f0fe40cba3490ccd089499a3 : SUCCESS in 4m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d424110b17d42d8853a6d6633871fdc : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d09827a666cb4f579abc330c97f80e63 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/944628101de847d9be79a87559d01d9d : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"19b42cb0b09bd4617655eaa5f7501b85a76066c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 10:55:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c1ae54b12093e6dc401fbb9aa6347bfbd2d0cc54","revisions":{"017dc075776c31104e58bb152cfed2f410cd1551":{"kind":"REWORK","_number":1,"created":"2020-11-03 03:49:55.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/44/761044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/44/761044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/44/761044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/44/761044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/44/761044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/44/761044/1"}}},"commit":{"parents":[{"commit":"a2e06fcfb9a606bd2609572999c810aae8ee65d2","subject":"Merge \"Cleanup for Refactor-error-messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/a2e06fcfb9a606bd2609572999c810aae8ee65d2"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2020-11-03 03:47:30.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2020-11-03 03:49:36.000000000","tz":480},"subject":"Remove six","message":"Remove six\n\nRemove six Replace the following items with Python 3 style code.\n- six.text_type\n- six.moves.urllib\n- six.PY3\n\nChange-Id: I8d2357e51d570e9e0513326248878fd0133144bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/017dc075776c31104e58bb152cfed2f410cd1551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/017dc075776c31104e58bb152cfed2f410cd1551"}]},"branch":"refs/heads/master"},"c1ae54b12093e6dc401fbb9aa6347bfbd2d0cc54":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-03 10:22:56.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/44/761044/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/44/761044/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/44/761044/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/44/761044/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/44/761044/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/44/761044/2"}}},"commit":{"parents":[{"commit":"a2e06fcfb9a606bd2609572999c810aae8ee65d2","subject":"Merge \"Cleanup for Refactor-error-messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/a2e06fcfb9a606bd2609572999c810aae8ee65d2"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2020-11-03 03:47:30.000000000","tz":480},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2020-11-03 10:21:10.000000000","tz":60},"subject":"Remove six","message":"Remove six\n\nRemove six Replace the following items with Python 3 style code.\n- six.text_type\n- six.moves.urllib\n- six.PY3\n\nChange-Id: I8d2357e51d570e9e0513326248878fd0133144bb\nStory: 2008305\nTask: 41203\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/c1ae54b12093e6dc401fbb9aa6347bfbd2d0cc54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/c1ae54b12093e6dc401fbb9aa6347bfbd2d0cc54"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
