)]}'
{"id":"openstack%2Fmonasca-ui~780195","triplet_id":"openstack%2Fmonasca-ui~master~Iddc99691aface0757ca8440711ae680799044901","project":"openstack/monasca-ui","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Iddc99691aface0757ca8440711ae680799044901","subject":"Imported Translations from Zanata","status":"MERGED","created":"2021-03-12 07:00:47.000000000","updated":"2021-03-18 12:43:56.000000000","submitted":"2021-03-18 12:43:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780195-zanata/translations","meta_rev_id":"d4cff8e4e83c486ed190df01dba1a0da3bad005a","_number":780195,"virtual_id_number":780195,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":17331,"name":"Koji Nakazono","email":"nakazono_0507@jp.fujitsu.com","username":"Koji_Nakazono"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 12:43:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17981,"name":"Haruki Yamanashi","email":"h.yamanashi@jp.fujitsu.com","username":"Haruki"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"recommended":{"_account_id":17331,"name":"Koji Nakazono","email":"nakazono_0507@jp.fujitsu.com","username":"Koji_Nakazono"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2021-03-18 12:08:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17331,"name":"Koji Nakazono","email":"nakazono_0507@jp.fujitsu.com","username":"Koji_Nakazono"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17981,"name":"Haruki Yamanashi","email":"h.yamanashi@jp.fujitsu.com","username":"Haruki"},{"value":2,"date":"2021-03-18 12:21:42.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":17331,"name":"Koji Nakazono","email":"nakazono_0507@jp.fujitsu.com","username":"Koji_Nakazono"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17981,"name":"Haruki Yamanashi","email":"h.yamanashi@jp.fujitsu.com","username":"Haruki"},{"value":1,"date":"2021-03-18 12:21:42.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17331,"name":"Koji Nakazono","email":"nakazono_0507@jp.fujitsu.com","username":"Koji_Nakazono"},{"_account_id":17981,"name":"Haruki Yamanashi","email":"h.yamanashi@jp.fujitsu.com","username":"Haruki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-12 07:11:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-12 08:04:45.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2021-03-12 09:17:47.000000000","updated_by":{"_account_id":17981,"name":"Haruki Yamanashi","email":"h.yamanashi@jp.fujitsu.com","username":"Haruki"},"reviewer":{"_account_id":17981,"name":"Haruki Yamanashi","email":"h.yamanashi@jp.fujitsu.com","username":"Haruki"},"state":"REVIEWER"},{"updated":"2021-03-12 09:17:47.000000000","updated_by":{"_account_id":17981,"name":"Haruki Yamanashi","email":"h.yamanashi@jp.fujitsu.com","username":"Haruki"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2021-03-12 09:17:47.000000000","updated_by":{"_account_id":17981,"name":"Haruki Yamanashi","email":"h.yamanashi@jp.fujitsu.com","username":"Haruki"},"reviewer":{"_account_id":17331,"name":"Koji Nakazono","email":"nakazono_0507@jp.fujitsu.com","username":"Koji_Nakazono"},"state":"REVIEWER"}],"messages":[{"id":"d030990f24f4941a4d1b6d0de592012eec14a0c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-12 07:00:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5406fe2201413bb335ee7cebd6ecf6235304bb2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 07:11:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0cda6e57b464806808ead7389fa3c25 : SUCCESS in 5m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97b3e231a025413598e25c1c630236e8 : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef03e167dbc44fbd9dea904a5fd8e0f8 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee32177b6da04687b2b89ec523186db4 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8760c6cf2e2943c29dfa42894c57b3f3 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0da5dfad233840d79b65a96cbab07e6a : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e37ffe0d74109558396ff3af489909fc3bcd597","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-03-12 08:04:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8a9c1da90190293623b876fbb0fea2dd02215291","author":{"_account_id":17981,"name":"Haruki Yamanashi","email":"h.yamanashi@jp.fujitsu.com","username":"Haruki"},"date":"2021-03-12 09:17:47.000000000","message":"Patch Set 1: Code-Review-1\n\nSorry, I mistook to translate L278. I fixed this line in Zanata and fix will be imported tomorrow so please wait to merge.","accounts_in_message":[],"_revision_number":1},{"id":"a3728d1999d65de1b0c41fc340b7a744dba9d673","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-03-12 09:27:11.000000000","message":"Removed Code-Review+2 by Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"5f5d336614460e1234380ee283243f1bc2794649","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-13 07:02:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1a605fc77ae686579238b31241ed90ddfae9e1d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 07:16:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a970cdeccb544c8789fdab28c384f13c : SUCCESS in 5m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e7e6a1e90be43d5b5c92d4227e003f6 : SUCCESS in 8m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/955559f5b7ee4ceb8a6f6b1c58a08a25 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/655d1d8a07a24bee9b7d052b62646edb : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af5ccd9f9a254821a1d02afe83207276 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a1febddfd554976ba24ddf631bd63fe : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f9cb9061307fa22166a243cd406a4bc34fee7d33","author":{"_account_id":17331,"name":"Koji Nakazono","email":"nakazono_0507@jp.fujitsu.com","username":"Koji_Nakazono"},"date":"2021-03-18 12:08:48.000000000","message":"Patch Set 2: Code-Review+1\n\nlooks good translation","accounts_in_message":[],"_revision_number":2},{"id":"2b696ef11b290c3a1c49828cdd965a931c6ec47a","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-03-18 12:21:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"904453a99f2ca5c6e677908670877f163b1578c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 12:30:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e76adecbd130d879a03785a30cfc3619bbdaea3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 12:43:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f7baee8a0844809b547b27b8ab1404c : SUCCESS in 6m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1df35ab6c06946bcb4a4b8486eeb8a27 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d949c72a176143608910e3cc395904ab : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8aef2f9f9bf0403d89bb150f643e31d7 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":2},{"id":"d4cff8e4e83c486ed190df01dba1a0da3bad005a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 12:43:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eb44a1031e473c32a2e5bdf7e495854632564294","revisions":{"b1d3019e898f64fd5e93cf4c5fd1c650d815d432":{"kind":"REWORK","_number":1,"created":"2021-03-12 07:00:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/780195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/95/780195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/95/780195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/95/780195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/95/780195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/95/780195/1"}}},"commit":{"parents":[{"commit":"eb069566f7cedbe83b2cc25d56c1d20390a02e7a","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/eb069566f7cedbe83b2cc25d56c1d20390a02e7a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 07:00:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-12 07:00:45.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Iddc99691aface0757ca8440711ae680799044901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/b1d3019e898f64fd5e93cf4c5fd1c650d815d432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/b1d3019e898f64fd5e93cf4c5fd1c650d815d432"}]},"branch":"refs/heads/master"},"eb44a1031e473c32a2e5bdf7e495854632564294":{"kind":"REWORK","_number":2,"created":"2021-03-13 07:02:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/780195/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/95/780195/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/95/780195/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/95/780195/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/95/780195/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/95/780195/2"}}},"commit":{"parents":[{"commit":"eb069566f7cedbe83b2cc25d56c1d20390a02e7a","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/eb069566f7cedbe83b2cc25d56c1d20390a02e7a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-13 07:02:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-13 07:02:46.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Iddc99691aface0757ca8440711ae680799044901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/eb44a1031e473c32a2e5bdf7e495854632564294"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/eb44a1031e473c32a2e5bdf7e495854632564294"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
