)]}'
{"id":"openstack%2Fmonasca-ui~809520","triplet_id":"openstack%2Fmonasca-ui~master~Ifacecc79608df3d052679779bc7a9f1338d2d2f6","project":"openstack/monasca-ui","branch":"master","hashtags":[],"change_id":"Ifacecc79608df3d052679779bc7a9f1338d2d2f6","subject":"Bump decorator version in lower-constraints.txt","status":"MERGED","created":"2021-09-17 04:39:21.000000000","updated":"2021-09-17 11:15:44.000000000","submitted":"2021-09-17 11:15:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809520","meta_rev_id":"add893e170c13d9d5ea00d5d28d10fdf8c78b92e","_number":809520,"virtual_id_number":809520,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-17 11:15:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-17 11:04:58.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-17 11:04:58.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 04:48:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-17 10:54:03.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"}],"messages":[{"id":"9e6ad69322a9915d03c566e803183c738df76497","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-17 04:39:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"107069fd5404ccbaf4e55429c59080407fa6cbdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 04:48:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/abdcd8305c524cad90b7da330c7af15c : FAILURE in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d8a2658fa6e47d49d1385063a6595a5 : SUCCESS in 5m 48s\n- horizon-tox-python3-django30 https://zuul.opendev.org/t/openstack/build/ab064ccce95a4a2b8a12c896f4ae7b97 : SUCCESS in 5m 50s (non-voting)\n- horizon-tox-python3-django31 https://zuul.opendev.org/t/openstack/build/8f72d5d596084da59bf30ec07f9410a9 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/659a738916f34ba78b49c1cd303bb532 : SUCCESS in 8m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c091e0bbd6f4d29947f3dcbd4b3130f : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8108df667b2f4186bc2ff060bbeff10e : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d0ce8c9bff44e0587f0ac781194fe19 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13797c75ff174ed5a9f45281e9262afd : SUCCESS in 5m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7eaeaf064b9281a49d2f1619b162cbc22fb7cf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-17 07:17:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5307e16cd31e2d6b553c1edc1e1e2e84b905c6ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 07:35:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1f921099494e4ac5b0814ebd6501ab19 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e29bfc441eca4e86bf856d1878a765ef : SUCCESS in 8m 33s\n- horizon-tox-python3-django30 https://zuul.opendev.org/t/openstack/build/3e3c65da84cc48208295cc5bfc7aab15 : SUCCESS in 7m 46s (non-voting)\n- horizon-tox-python3-django31 https://zuul.opendev.org/t/openstack/build/db116e7564f6441da5fc9c7f71db3941 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/effc7107df4341e0aa5450a42a76ee4f : SUCCESS in 9m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45b87880e48d44c29dec356ef281ba52 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71849d4c163a4d82bbfdd74d3f879c56 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62d8f7667cf043b5bd8dbeeb70d16df5 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ef01decbcc54d6899979bf7a89b5b40 : SUCCESS in 7m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"68adc2568f54ec5725dbbabaa2942262fec1606b","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-09-17 11:04:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks for the fix","accounts_in_message":[],"_revision_number":2},{"id":"47c1d6fdbd637084dc22157bbc48d028f0a367e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 11:05:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"71132d9da26279490f1b222e26e5473643ed875f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 11:15:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/111c6b5dddc54767a9e4841882983dcf : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57c4e7b865b34c6c9d98df8585049b78 : SUCCESS in 4m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a403e29aeff43bd8dae6e49d1f75b7e : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6a7f0a40dc4eb1aa1e7624b81f2556 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fcd81a0b8a74c59936e994cfc35ab33 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"add893e170c13d9d5ea00d5d28d10fdf8c78b92e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 11:15:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ced97891391fb41890124f36bc90a3b6ef1634b4","revisions":{"10a61a0dabd0f968b14c95320d520fead39b93f5":{"kind":"REWORK","_number":1,"created":"2021-09-17 04:39:21.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/20/809520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/20/809520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/20/809520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/20/809520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/20/809520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/20/809520/1"}}},"commit":{"parents":[{"commit":"9014bce3de6fed79f85b79ecf30031e6af7b7356","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/9014bce3de6fed79f85b79ecf30031e6af7b7356"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-17 04:38:36.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-17 04:39:18.000000000","tz":330},"subject":"Bump decorator version in lower-constraints.txt","message":"Bump decorator version in lower-constraints.txt\n\nThis patch update decorator version to 4.4.2 in\nlower-constraints.txt to fix the lower-constraints job.\n\nChange-Id: Ifacecc79608df3d052679779bc7a9f1338d2d2f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/10a61a0dabd0f968b14c95320d520fead39b93f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/10a61a0dabd0f968b14c95320d520fead39b93f5"}]},"branch":"refs/heads/master"},"ced97891391fb41890124f36bc90a3b6ef1634b4":{"kind":"REWORK","_number":2,"created":"2021-09-17 07:17:43.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/20/809520/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/20/809520/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/20/809520/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/20/809520/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/20/809520/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/20/809520/2"}}},"commit":{"parents":[{"commit":"9014bce3de6fed79f85b79ecf30031e6af7b7356","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/9014bce3de6fed79f85b79ecf30031e6af7b7356"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-17 04:38:36.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-17 07:17:19.000000000","tz":330},"subject":"Bump decorator version in lower-constraints.txt","message":"Bump decorator version in lower-constraints.txt\n\nThis patch update decorator version to 4.4.2 in\nlower-constraints.txt to fix the lower-constraints job.\nIt also update horizon version in requiremnets.txt to\nsync with horizon version in lower-constraints.txt.\n\nChange-Id: Ifacecc79608df3d052679779bc7a9f1338d2d2f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/ced97891391fb41890124f36bc90a3b6ef1634b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/ced97891391fb41890124f36bc90a3b6ef1634b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
