)]}'
{"id":"openstack%2Fmonasca-ui~858188","triplet_id":"openstack%2Fmonasca-ui~master~I4a7d2b7673a1f42ca3cc127d9d163f0b002847ea","project":"openstack/monasca-ui","branch":"master","topic":"add-antelope-python-jobtemplates","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2025-07-20 21:34:21.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"13114":{"account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"last_update":"2025-07-21 09:50:24.000000000","reason":"\u003cGERRIT_ACCOUNT_13114\u003e replied on the change","reason_account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-07-21 09:51:12.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I4a7d2b7673a1f42ca3cc127d9d163f0b002847ea","subject":"Switch to 2023.1 Python3 unit tests and generic template name","status":"MERGED","created":"2022-09-16 15:24:12.000000000","updated":"2025-07-21 10:00:02.000000000","submitted":"2025-07-21 10:00:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858188-add-antelope-python-jobtemplates","meta_rev_id":"bb1a25ad3bfcabda7e10288d550d89aafd6f9c0d","_number":858188,"virtual_id_number":858188,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-21 10:00:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2025-07-21 09:50:24.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-21 08:43:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2025-07-21 09:50:24.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-16 15:39:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-23 03:22:03.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2025-07-20 17:51:40.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-07-20 21:34:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-07-20 21:34:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"state":"REVIEWER"}],"messages":[{"id":"cbed3238374f3c96ee51ebda5ef0130cbb011b9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-09-16 15:24:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b02fa63dc30dd2059d624aca4f20431182f32f8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 15:39:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/815eb72ed1f442d5b49eaaface31aa08\n\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9863a881ce1045a5975b1f245022bd9f : SUCCESS in 6m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/970599d7bd574f61bb8fcfa51d971e9f : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f560c0713ad546f0a185abc3dcd0e775 : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/306d5c44a3564574a19a983598f43919 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"addc280b5a2b965f0cdbc885213eb3668dcf93dc","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-11-23 03:22:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"026b41f5d1ee6c66fa10999da51f43a2bb69e5de","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-09 16:36:26.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"98bc67b280d39fd4ccceb94098e5fa6d1ad19910","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 16:36:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d6a3d2f6525c5563d34ea6250d0f8f55db6c256","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 16:44:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c134b2cdf38e4dc1a43fbee39f2bf841\n\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48494bbe1b6144debc5f8662bb879e2f : FAILURE in 1m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5424da9fc3df4157987137929ada272a : FAILURE in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dc4afbdb70741159565050f90e00b90 : FAILURE in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02a47b2e46de49cea3bcf46486ace4dc : FAILURE in 2m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6aa2c602f104edfeffd670240df77e25d2c7b1b0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-20 17:51:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"802bf98ea0fdaecf43739a54d572688ded943cab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-20 18:07:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d545577503e4452f95879a95ee988010\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/707803be02d74b0581e82f3a50bf813d : FAILURE in 1m 57s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/1a0655fd040444a7a357dd8acda7b8a6 : RETRY_LIMIT in 2m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/858346306522498e8fef4ebc8a63f777 : FAILURE in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53cdec21642d48278dcb9bff7952d7a2 : FAILURE in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63779b192ccc409bad2dc7a582ef31b9 : FAILURE in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a6f4e2fa25fc4e8d8c975ca7e1614d30 : FAILURE in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b44b1a4bd92dea24fe4048491e000932f3fa8f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-20 21:15:43.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fd0899d8dfd273dfe9b7f2c97324fbdcdf0ae424","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-20 21:15:59.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b68c9296654564ec9576600c8d2cbb456ba83252","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-20 21:16:11.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ff95b39e4b9e8881e27bc911fce532f42266095c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-20 21:16:33.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"147c9cccb76edbc6e3c9ac5afafbad5e445d558e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-20 21:30:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c105a4e9cce647bb8adc606c006f4f7a\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b25fef36df4f4b0dbb25f6e019f45c7a : SUCCESS in 2m 10s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/e95c9ca0887c402c992f8e31b5e68ded : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a384263bbeb44d6950859f73e716357 : SUCCESS in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a3aa84e80224e43b7bf0c808654b25b : SUCCESS in 1m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/657933e1d2a345cda910bde5a1424826 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/601b7572e2b942a98bd20e6d33900fae : SUCCESS in 3m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c517042d48cef1dbd964767317e734a347379796","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-20 21:34:18.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"e730d80fb13b730a14b49a01a76576f2eb33c81a","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-07-21 03:11:29.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1182845f26af4b6b8320f9a728b56da48f78b00c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-21 08:32:25.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"134baea6214a1b05cad211f13d768a4ed70fc787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 08:43:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79d158f0d9614726812990b88d364d9b\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1c6e72e3a1ef439a8ec30823e87e724d : SUCCESS in 5m 33s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/82b5b03899ef47f5958a338d4e898360 : SUCCESS in 4m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3ad9a79e02547d5bc1a0e64cd2cb1d0 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/996fcadd23c1489f96ef38d8b1e0d853 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de5c3bdc82a545a7ba362347439396dc : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ec74180ba2e44f0a9fa6ec7f277f55a : SUCCESS in 8m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7b188ffa836b01c9d7ef20c0a0656039357bb157","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-21 08:43:48.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dabccf943e65aa42b25bab8013b4a79926255d3e","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-07-21 09:50:24.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ca601a1164c15525040e4976fc83724e5f04e129","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 09:50:50.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e9b09312a9d4354f4bf2132d2cf29529c74edd14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 10:00:02.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e480912cc2f642e18d5b982d5380dadb\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/358bc8b14b5b4d028fd4aa0eff40190f : SUCCESS in 4m 28s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/e8c83ff2d40143b9ac7b494389a017aa : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/778a33c50ee1491ebc5cd246a72df720 : SUCCESS in 1m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39e48dc6f2a74213b471fd46eabeb1b2 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":6},{"id":"bb1a25ad3bfcabda7e10288d550d89aafd6f9c0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 10:00:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2317d95a9b6415bc0169677a430f7ede91e6f357","revisions":{"1a2096a61ccf6c15983e2f52963855cae6a03367":{"kind":"REWORK","_number":1,"created":"2022-09-16 15:24:12.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/88/858188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/88/858188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/1"}}},"commit":{"parents":[{"commit":"097d12aee254ab8bd89b87f3e7463a996e6813d9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/097d12aee254ab8bd89b87f3e7463a996e6813d9"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-16 15:24:09.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-16 15:24:09.000000000","tz":0},"subject":"Switch to 2023.1 Python3 unit tests and generic template name","message":"Switch to 2023.1 Python3 unit tests and generic template name\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for antelope. Also,\nupdating the template name to generic one.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I4a7d2b7673a1f42ca3cc127d9d163f0b002847ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/1a2096a61ccf6c15983e2f52963855cae6a03367"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/1a2096a61ccf6c15983e2f52963855cae6a03367"}]},"branch":"refs/heads/master"},"cbf66af420608527f81dbd1a31c8813bb7950d01":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-07-20 21:15:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/88/858188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/88/858188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/2"}}},"commit":{"parents":[{"commit":"097d12aee254ab8bd89b87f3e7463a996e6813d9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/097d12aee254ab8bd89b87f3e7463a996e6813d9"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-16 15:24:09.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-20 21:15:43.000000000","tz":0},"subject":"Switch to 2023.1 Python3 unit tests and generic template name","message":"Switch to 2023.1 Python3 unit tests and generic template name\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for antelope. Also,\nupdating the template name to generic one.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I4a7d2b7673a1f42ca3cc127d9d163f0b002847ea\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/cbf66af420608527f81dbd1a31c8813bb7950d01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/cbf66af420608527f81dbd1a31c8813bb7950d01"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a300711853fb7c6fc9add304aa9cd05d92dfabec":{"kind":"REWORK","_number":3,"created":"2025-07-20 21:15:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/88/858188/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/88/858188/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/3"}}},"commit":{"parents":[{"commit":"097d12aee254ab8bd89b87f3e7463a996e6813d9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/097d12aee254ab8bd89b87f3e7463a996e6813d9"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-16 15:24:09.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-20 21:15:55.000000000","tz":0},"subject":"Switch to 2023.1 Python3 unit tests and generic template name","message":"Switch to 2023.1 Python3 unit tests and generic template name\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for antelope. Also,\nupdating the template name to generic one.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I4a7d2b7673a1f42ca3cc127d9d163f0b002847ea\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/a300711853fb7c6fc9add304aa9cd05d92dfabec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/a300711853fb7c6fc9add304aa9cd05d92dfabec"}]},"branch":"refs/heads/master"},"8dd84a26937de6f613316972f679d8d4ebf01c3e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-07-20 21:16:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/88/858188/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/88/858188/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/4"}}},"commit":{"parents":[{"commit":"de5e83efd967f94142d3b5f6fba86b5aab72d7e7","subject":"Fix for various problems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/de5e83efd967f94142d3b5f6fba86b5aab72d7e7"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-16 15:24:09.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-20 21:16:11.000000000","tz":0},"subject":"Switch to 2023.1 Python3 unit tests and generic template name","message":"Switch to 2023.1 Python3 unit tests and generic template name\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for antelope. Also,\nupdating the template name to generic one.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I4a7d2b7673a1f42ca3cc127d9d163f0b002847ea\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/8dd84a26937de6f613316972f679d8d4ebf01c3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/8dd84a26937de6f613316972f679d8d4ebf01c3e"}]},"branch":"refs/heads/master","description":"Rebase"},"12f5ecd3d02175f7180749c8efa821d80fbe1356":{"kind":"REWORK","_number":5,"created":"2025-07-20 21:16:33.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/88/858188/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/88/858188/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/5"}}},"commit":{"parents":[{"commit":"de5e83efd967f94142d3b5f6fba86b5aab72d7e7","subject":"Fix for various problems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/de5e83efd967f94142d3b5f6fba86b5aab72d7e7"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-16 15:24:09.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-20 21:16:33.000000000","tz":0},"subject":"Switch to 2023.1 Python3 unit tests and generic template name","message":"Switch to 2023.1 Python3 unit tests and generic template name\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for antelope. Also,\nupdating the template name to generic one.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I4a7d2b7673a1f42ca3cc127d9d163f0b002847ea\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/12f5ecd3d02175f7180749c8efa821d80fbe1356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/12f5ecd3d02175f7180749c8efa821d80fbe1356"}]},"branch":"refs/heads/master"},"2317d95a9b6415bc0169677a430f7ede91e6f357":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-07-21 08:32:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/88/858188/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/88/858188/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/88/858188/6"}}},"commit":{"parents":[{"commit":"906065ca7030779d42bf051aebd95ae272b319c5","subject":"Fix for various problems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/906065ca7030779d42bf051aebd95ae272b319c5"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-16 15:24:09.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-21 08:32:25.000000000","tz":0},"subject":"Switch to 2023.1 Python3 unit tests and generic template name","message":"Switch to 2023.1 Python3 unit tests and generic template name\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for antelope. Also,\nupdating the template name to generic one.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I4a7d2b7673a1f42ca3cc127d9d163f0b002847ea\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/2317d95a9b6415bc0169677a430f7ede91e6f357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/2317d95a9b6415bc0169677a430f7ede91e6f357"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
