)]}'
{"id":"openstack%2Fmonasca-ui~954838","triplet_id":"openstack%2Fmonasca-ui~master~I9fd014fc273dd8a4c241e8f4dfd464d0f6fe8546","project":"openstack/monasca-ui","branch":"master","topic":"zanata/translations","attention_set":{},"removed_from_attention_set":{"13114":{"account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"last_update":"2025-07-21 08:54:22.000000000","reason":"\u003cGERRIT_ACCOUNT_13114\u003e replied on the change","reason_account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}}},"hashtags":[],"change_id":"I9fd014fc273dd8a4c241e8f4dfd464d0f6fe8546","subject":"Imported Translations from Zanata","status":"MERGED","created":"2025-07-12 03:14:05.000000000","updated":"2025-07-21 09:08:36.000000000","submitted":"2025-07-21 09:08:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954838-zanata/translations","meta_rev_id":"f3ea266ddd325598c506af8354d6f50ce259e01b","_number":954838,"virtual_id_number":954838,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-21 09:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":2,"date":"2025-07-21 08:54:22.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-21 08:40:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":1,"date":"2025-07-21 08:54:22.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-12 03:21:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-20 21:19:00.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-07-20 21:19:00.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"state":"REVIEWER"}],"messages":[{"id":"52511ff3585df0ece497005efae1646b908179be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-07-12 03:14:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc45032807cfa0aae4b46a0b27ccd9a112459eb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 03:21:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cbfd84cccd642d9b876796d96eb85a0\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e55ad9deb3eb4312997f9e0b4cea6e30 : FAILURE in 2m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8146dce14bc040108b2d50c748514373 : FAILURE in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2f1203c64b442b9b7725f54ff2317b0 : FAILURE in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb956149321844ba8235255ea5ac9db1 : FAILURE in 2m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc6f46f53c4a4d88a23240b031a3b405 : FAILURE in 2m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbd94035aebe48328b7cc99a40921df1 : FAILURE in 2m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"738ec3f122307e841d87c500401a21676fcd5fc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-20 21:13:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"678c443b7eee971a6fa92613f0183fa1c920fb63","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-20 21:19:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b4a8a6cbb16ecf364182847246720952ab667935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-20 21:22:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f1f96e03ce6408a9877d42adb784758\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2e05d42945445aea380473f3d9d592a : SUCCESS in 3m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56660742cc0c45fca9a708ffbb56e744 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":2},{"id":"aa5ef11ec09467cbad2648b074b376238f71ca92","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-07-21 03:12:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2452a088b46e167ea929fba3bec79870862a8ad5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-21 08:28:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4d615c70790b54b9fac390bdf34cda72dff3885","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-21 08:31:51.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":3},{"id":"20f114b52c27ca6963d01f73d06c35a50420cab3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 08:39:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffbc262880d8414e885f76c94c05b04f\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d618d1588764b678d518842beeb2d0b : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba512d3259934b1eacc8d81a1c9ca38f : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":3},{"id":"97beda02750863ca97ca2a286978c711c0045ffb","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-21 08:40:13.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a7fbe56a5be7ea577dc50a41dfde719a1ac5b40c","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-07-21 08:54:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3fc3053f8917a34de501ea56d2e1e4c3e41b180f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 08:54:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0297c6b8be18be9ed6f23056a5a3503d7018c5f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 09:08:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b2af98eb8c94d4ba2b1ddc3279381e4\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46f5aaa540fb455da64f302f601123aa : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":3},{"id":"f3ea266ddd325598c506af8354d6f50ce259e01b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 09:08:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"aac8a8a33344eb36b85a340af806459d81db80cf","revisions":{"efc9d90116526b306e50b1d5ee978e316ba96ffe":{"kind":"REWORK","_number":1,"created":"2025-07-12 03:14:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/954838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/38/954838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/38/954838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/38/954838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/38/954838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/38/954838/1"}}},"commit":{"parents":[{"commit":"097d12aee254ab8bd89b87f3e7463a996e6813d9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/097d12aee254ab8bd89b87f3e7463a996e6813d9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-12 03:14:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-12 03:14:02.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I9fd014fc273dd8a4c241e8f4dfd464d0f6fe8546\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/openstack-zuul-jobs:roles/prepare-zanata-client/files/common_translation_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/efc9d90116526b306e50b1d5ee978e316ba96ffe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/efc9d90116526b306e50b1d5ee978e316ba96ffe"}]},"branch":"refs/heads/master"},"42b1ba3a95f9c6c67b463d1b385d07bb2d4e4e2f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-20 21:13:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/38/954838/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/38/954838/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/38/954838/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/38/954838/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/38/954838/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/38/954838/2"}}},"commit":{"parents":[{"commit":"de5e83efd967f94142d3b5f6fba86b5aab72d7e7","subject":"Fix for various problems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/de5e83efd967f94142d3b5f6fba86b5aab72d7e7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-12 03:14:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-20 21:13:57.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I9fd014fc273dd8a4c241e8f4dfd464d0f6fe8546\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/openstack-zuul-jobs:roles/prepare-zanata-client/files/common_translation_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/42b1ba3a95f9c6c67b463d1b385d07bb2d4e4e2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/42b1ba3a95f9c6c67b463d1b385d07bb2d4e4e2f"}]},"branch":"refs/heads/master","description":"Rebase"},"aac8a8a33344eb36b85a340af806459d81db80cf":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-07-21 08:31:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/38/954838/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ui","ref":"refs/changes/38/954838/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/38/954838/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/38/954838/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ui refs/changes/38/954838/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ui refs/changes/38/954838/3"}}},"commit":{"parents":[{"commit":"906065ca7030779d42bf051aebd95ae272b319c5","subject":"Fix for various problems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/906065ca7030779d42bf051aebd95ae272b319c5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-12 03:14:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-21 08:31:51.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I9fd014fc273dd8a4c241e8f4dfd464d0f6fe8546\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/openstack-zuul-jobs:roles/prepare-zanata-client/files/common_translation_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/aac8a8a33344eb36b85a340af806459d81db80cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ui/commit/aac8a8a33344eb36b85a340af806459d81db80cf"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
