)]}'
{"id":"openstack%2Fmox3~593566","triplet_id":"openstack%2Fmox3~master~I54f99e7e29c7b3727db98bdaf8bc168da3890a1d","project":"openstack/mox3","branch":"master","hashtags":[],"change_id":"I54f99e7e29c7b3727db98bdaf8bc168da3890a1d","subject":"Remove flake8 from the test-requirements","status":"MERGED","created":"2018-08-20 10:03:19.000000000","updated":"2019-09-06 03:16:14.000000000","submitted":"2019-09-06 03:14:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"593566-1567739649567-7e60562c","meta_rev_id":"d7c066823791bb506540fa4dd64d9099451091b8","_number":593566,"virtual_id_number":593566,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-09-06 03:14:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-08-13 15:46:38.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-09-06 02:53:27.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-09-06 02:53:27.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-20 13:08:16.000000000","updated_by":{"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},"reviewer":{"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},"state":"REVIEWER"},{"updated":"2018-10-15 15:40:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-08-13 15:46:38.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-09-06 02:53:27.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-09-06 03:14:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f570d9a9c83854a98c8902d18f191722ba7156f","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2018-08-20 10:03:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d8b4bc4c883924a8d5bf8fb0330797dad2cfdab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-20 10:13:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/66/593566/1/check/openstack-tox-pep8/26760a0/ : SUCCESS in 8m 36s\n- openstack-tox-py27 http://logs.openstack.org/66/593566/1/check/openstack-tox-py27/c991c5c/ : SUCCESS in 5m 22s\n- openstack-tox-py35 http://logs.openstack.org/66/593566/1/check/openstack-tox-py35/b348f23/ : SUCCESS in 5m 18s\n- openstack-tox-docs http://logs.openstack.org/66/593566/1/check/openstack-tox-docs/1834d9d/html/ : SUCCESS in 4m 00s\n- requirements-check http://logs.openstack.org/66/593566/1/check/requirements-check/efb4a33/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/66/593566/1/check/openstack-tox-py36/28a227f/ : SUCCESS in 4m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/66/593566/1/check/openstack-tox-lower-constraints/1e5012e/ : SUCCESS in 9m 12s","accounts_in_message":[],"_revision_number":1},{"id":"278e7c37f84c9bc343c2d0d56d5cf892e3edc5f8","author":{"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},"date":"2018-08-20 13:08:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"19f3e7b05bff87061464d4ce1af09e90eb18e870","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2018-10-03 17:57:01.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e191cf0acb21ab41d2b1ef862bee25cec1862177","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-10-15 15:40:34.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI can go either way on this. On one hand, what you say is correct - flake8 isn\u0027t needed for anything else. On the other hand, every other project does things this way presumably for simplicity\u0027s sake. In fact, in nova we take advantage of this and use a shared environment for pretty much all non-Python versioned targets [1].\n\nIf you really want this, I\u0027m happy to oblige (though I think we should also do the opposite - see inline). If it\u0027s not that important though, I\u0027d rather we keep things the way they are. Your call :)\n\n[1] https://github.com/openstack/nova/blob/396156eb/tox.ini#L58","accounts_in_message":[],"_revision_number":1},{"id":"c515500125e004eefb198235e6583becdeee1a80","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-08-12 16:46:39.000000000","message":"Patch Set 1: Code-Review-1\n\nGiven that this has been lingering for a long time and I still have some reservations about why we would do it, I\u0027m going to -1. At the very least I would like to see the commit message updated to explain what benefit this brings as it isn\u0027t clear to me that either of the things mentioned are valid.\n\nAlso, since this is a common pattern across OpenStack, I\u0027m curious if the plan is to make this change everywhere. Removing flake8 from mox3 doesn\u0027t buy us much since mox isn\u0027t used in most projects anymore.","accounts_in_message":[],"_revision_number":1},{"id":"0c01bce991b0c5912dce78d170573b53fcc6922e","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-13 15:34:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"758bb209f4c2e7916952475fc20557630c6947cb","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-13 15:34:11.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2ce2143b480e7f1b15654acc98c964ad7240745e","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-08-13 15:46:38.000000000","message":"Patch Set 2: Code-Review+2\n\nOkay, this makes more sense to me now. Thanks for the explanation.\n\nSince we don\u0027t use shared venvs I\u0027m good with making this change here.","accounts_in_message":[],"_revision_number":2},{"id":"d765f6c36749d4b389e5e982af3fe4063d6535ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 15:46:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7527ea58c8bc4782a486243c0ce292c5 : SUCCESS in 2m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0a555f11a16481398e38b0d29caede1 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d22f89ca5eb64442a737e530d40cfd71 : SUCCESS in 2m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c73079587b54bfbafcbb59510bc942f : SUCCESS in 2m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4a4c69a569444789247aaf634f99ad3 : SUCCESS in 3m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1433b3f4e8943568e4acb46b3a64677 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f0bad98fac34bb5a2b4392a952fd10c : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"ac28c77066ad92a0381adcae7b8777ce55bb3dc4","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-09-06 02:53:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5d67ce70090ea5261f7d02544ff8c5b64df79d52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 02:53:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0c3281a80ef67dbf3a694f94add5a400edaaee7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 03:14:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bd112c2f1c0042e38ef86e28069d2ada : SUCCESS in 1m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e054fdb79d64ce28d16b22d99b0aa19 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd804151532f4fd5bd9cab25523dbce1 : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c3323eae227f4f4da7487bad1ca97094 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a637137cde3d4f47b6dd373319981d04 : SUCCESS in 2m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7245144db0b489fad1d369437530ca8 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba15718c5a504fd5ab13c8c6953e9e18 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"5b3343db2f8ac9c0d4755fb65f47f5ccc4672854","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 03:14:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d7c066823791bb506540fa4dd64d9099451091b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 03:16:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c431241272f54374a7b4935da95d1ac4 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a7706ca345cebcb5e8443104e15e4656f742e327","revisions":{"9e27c33eea7f1e38bc10bc8a790a133f2b5d3296":{"kind":"REWORK","_number":1,"created":"2018-08-20 10:03:19.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/66/593566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mox3","ref":"refs/changes/66/593566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/66/593566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/66/593566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/66/593566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mox3 refs/changes/66/593566/1"}}},"commit":{"parents":[{"commit":"34a010e1681d66783319bfa18aa7dd6dbe6adc6b","subject":"add python 3.6 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/34a010e1681d66783319bfa18aa7dd6dbe6adc6b"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2018-08-20 10:02:08.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2018-08-20 10:02:08.000000000","tz":120},"subject":"Remove flake8 from the test-requirements","message":"Remove flake8 from the test-requirements\n\nthe linter is a blacklisted dependency, so we don\u0027t want to\npull it in for unit tests, as that causes any setup.py driven\nbuild/install process to require flake8, which is not actually needed.\n\nAlso update to the newer flake8 2.6.x which is needed by current\nhacking (which causes no fallout so should be fine).\n\nChange-Id: I54f99e7e29c7b3727db98bdaf8bc168da3890a1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/9e27c33eea7f1e38bc10bc8a790a133f2b5d3296"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/9e27c33eea7f1e38bc10bc8a790a133f2b5d3296"}]},"branch":"refs/heads/master"},"a7706ca345cebcb5e8443104e15e4656f742e327":{"kind":"REWORK","_number":2,"created":"2019-08-13 15:34:09.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/66/593566/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mox3","ref":"refs/changes/66/593566/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/66/593566/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/66/593566/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/66/593566/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mox3 refs/changes/66/593566/2"}}},"commit":{"parents":[{"commit":"34a010e1681d66783319bfa18aa7dd6dbe6adc6b","subject":"add python 3.6 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/34a010e1681d66783319bfa18aa7dd6dbe6adc6b"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2018-08-20 10:02:08.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-08-13 15:32:14.000000000","tz":120},"subject":"Remove flake8 from the test-requirements","message":"Remove flake8 from the test-requirements\n\nflake8 is a linter so we don\u0027t need to pull in for the unit\ntests or other tox environments.\n\nAlso update to the newer flake8 2.6.x which is the version\nchosen by current hacking release (which causes no fallout so should be fine).\nAlso this is the first version that is compatible being run\nunder python 3.6/7.\n\nChange-Id: I54f99e7e29c7b3727db98bdaf8bc168da3890a1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/a7706ca345cebcb5e8443104e15e4656f742e327"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/a7706ca345cebcb5e8443104e15e4656f742e327"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
