)]}'
{"id":"openstack%2Fmox3~652433","triplet_id":"openstack%2Fmox3~master~I097138c3a4262a844819478d0f8a94180d16fa35","project":"openstack/mox3","branch":"master","topic":"dropping-py35-testing","hashtags":[],"change_id":"I097138c3a4262a844819478d0f8a94180d16fa35","subject":"Dropping the py35 testing","status":"MERGED","created":"2019-04-15 00:52:56.000000000","updated":"2019-05-03 16:46:51.000000000","submitted":"2019-05-03 16:46:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652433-1556902011200-13d10e42","meta_rev_id":"356f5319cd31e61cfa67f11fc567fe5dced3a4c0","_number":652433,"virtual_id_number":652433,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-05-03 16:46:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-05-03 16:26:10.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-05-03 16:26:10.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-03 16:26:10.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-03 16:46:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4fa8f5306763b35282b4d5aa37a4fa2c073e4201","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-15 00:52:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4018cbced48afcd51c1ef184711300bd7d03f569","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 00:59:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/33/652433/1/check/test-release-openstack/59f189f/ : SUCCESS in 2m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/33/652433/1/check/openstack-tox-lower-constraints/5b06586/ : SUCCESS in 5m 12s\n- openstack-tox-pep8 http://logs.openstack.org/33/652433/1/check/openstack-tox-pep8/a4377f9/ : SUCCESS in 3m 48s\n- openstack-tox-py27 http://logs.openstack.org/33/652433/1/check/openstack-tox-py27/b4112c1/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/33/652433/1/check/openstack-tox-py36/62255ef/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/33/652433/1/check/openstack-tox-docs/099d66b/html/ : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d442252a16256706f558d0bb861a846b474f51ed","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-15 18:16:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b82c0fce5f9402be2ec122f7d2b5311da9fb1476","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 18:35:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/33/652433/2/check/test-release-openstack/fa28b6e/ : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/33/652433/2/check/openstack-tox-lower-constraints/67b81fb/ : SUCCESS in 7m 16s\n- openstack-tox-pep8 http://logs.openstack.org/33/652433/2/check/openstack-tox-pep8/a72da05/ : SUCCESS in 5m 35s\n- openstack-tox-py27 http://logs.openstack.org/33/652433/2/check/openstack-tox-py27/965fe6a/ : SUCCESS in 6m 06s\n- openstack-tox-py36 http://logs.openstack.org/33/652433/2/check/openstack-tox-py36/d6ee4f7/ : SUCCESS in 6m 10s\n- openstack-tox-docs http://logs.openstack.org/33/652433/2/check/openstack-tox-docs/afcf7fd/html/ : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":2},{"id":"a55cf5a7670249e26607d6b24a6c5a76dd752a54","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 16:26:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nTrivial change. Fast approving","accounts_in_message":[],"_revision_number":2},{"id":"268974985f2be70143ec692f2667271627d1e451","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:26:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"232527e1cae96646f2901cd4bb83dc35fcd302e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:46:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/33/652433/2/gate/test-release-openstack/45a9d62/ : SUCCESS in 5m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/33/652433/2/gate/openstack-tox-lower-constraints/45f0fc6/ : SUCCESS in 5m 02s\n- openstack-tox-pep8 http://logs.openstack.org/33/652433/2/gate/openstack-tox-pep8/c4292c2/ : SUCCESS in 6m 13s\n- openstack-tox-py27 http://logs.openstack.org/33/652433/2/gate/openstack-tox-py27/000ac72/ : SUCCESS in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/33/652433/2/gate/openstack-tox-py36/8fe4a91/ : SUCCESS in 6m 11s\n- openstack-tox-docs http://logs.openstack.org/33/652433/2/gate/openstack-tox-docs/d7d34a8/html/ : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":2},{"id":"02aaaab680bd686c2885171702ced5623694a3f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:46:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"71071cc2172e5a1bd9c42e33b5f06a2daf891888","revisions":{"b7fe28eb8586b901fc7d21c55f46894f5d5d1467":{"kind":"REWORK","_number":1,"created":"2019-04-15 00:52:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/33/652433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mox3","ref":"refs/changes/33/652433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/33/652433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/33/652433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/33/652433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mox3 refs/changes/33/652433/1"}}},"commit":{"parents":[{"commit":"543e2cc3772f0ec856de314c1cd0088baa341f22","subject":"Use template for lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/543e2cc3772f0ec856de314c1cd0088baa341f22"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 00:52:38.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 00:52:54.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I097138c3a4262a844819478d0f8a94180d16fa35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/b7fe28eb8586b901fc7d21c55f46894f5d5d1467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/b7fe28eb8586b901fc7d21c55f46894f5d5d1467"}]},"branch":"refs/heads/master"},"71071cc2172e5a1bd9c42e33b5f06a2daf891888":{"kind":"REWORK","_number":2,"created":"2019-04-15 18:16:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/33/652433/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mox3","ref":"refs/changes/33/652433/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/33/652433/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/33/652433/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/33/652433/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mox3 refs/changes/33/652433/2"}}},"commit":{"parents":[{"commit":"543e2cc3772f0ec856de314c1cd0088baa341f22","subject":"Use template for lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/543e2cc3772f0ec856de314c1cd0088baa341f22"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 00:52:38.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 18:16:35.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I097138c3a4262a844819478d0f8a94180d16fa35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/71071cc2172e5a1bd9c42e33b5f06a2daf891888"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/71071cc2172e5a1bd9c42e33b5f06a2daf891888"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
