)]}'
{"id":"openstack%2Fmox3~655024","triplet_id":"openstack%2Fmox3~master~I613b5676676215c6f80d309799a2efd715894ee7","project":"openstack/mox3","branch":"master","hashtags":[],"change_id":"I613b5676676215c6f80d309799a2efd715894ee7","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 11:38:39.000000000","updated":"2019-07-19 16:58:36.000000000","submitted":"2019-05-03 18:44:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655024-1556909048613-13d10e42","meta_rev_id":"afa85dc796edd69728f4a76f13173572749f965d","_number":655024,"virtual_id_number":655024,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-05-03 18:44:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-05-03 18:32:14.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-05-02 15:46:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-05-03 16:39:22.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-04-24 10:04:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-05-03 18:32:14.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 10:04:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-05-02 15:46:58.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-05-03 16:39:22.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-03 18:32:14.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-05-03 18:44:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6d6a0a0fa7b1e9ad0f88d34eecabf4fa5b0f086","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 11:38:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2fb64eaf713478e35230d7c07de92c339908791","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:23:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/24/655024/1/check/test-release-openstack/e44efaf/ : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/24/655024/1/check/openstack-tox-lower-constraints/ffeb78a/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/24/655024/1/check/openstack-tox-pep8/596ee35/ : SUCCESS in 6m 14s\n- openstack-tox-py27 http://logs.openstack.org/24/655024/1/check/openstack-tox-py27/68d34d5/ : SUCCESS in 6m 25s\n- openstack-tox-py35 http://logs.openstack.org/24/655024/1/check/openstack-tox-py35/797fefc/ : SUCCESS in 5m 50s\n- openstack-tox-py36 http://logs.openstack.org/24/655024/1/check/openstack-tox-py36/016c658/ : SUCCESS in 6m 35s\n- openstack-tox-docs http://logs.openstack.org/24/655024/1/check/openstack-tox-docs/11b5cd3/html/ : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d00a090bf14c95032676b2ef788eba2dad5ba128","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 13:34:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b343bb0059973bb545e44ba7ecf56a27319335b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:58:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/24/655024/2/check/test-release-openstack/87a7a1b/ : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/24/655024/2/check/openstack-tox-lower-constraints/4a1558b/ : SUCCESS in 7m 20s\n- openstack-tox-pep8 http://logs.openstack.org/24/655024/2/check/openstack-tox-pep8/8b89b03/ : SUCCESS in 6m 08s\n- openstack-tox-py27 http://logs.openstack.org/24/655024/2/check/openstack-tox-py27/8d5efc2/ : SUCCESS in 6m 10s\n- openstack-tox-py35 http://logs.openstack.org/24/655024/2/check/openstack-tox-py35/9bd2432/ : SUCCESS in 5m 33s\n- openstack-tox-py36 http://logs.openstack.org/24/655024/2/check/openstack-tox-py36/2d7d603/ : SUCCESS in 7m 09s\n- openstack-tox-docs http://logs.openstack.org/24/655024/2/check/openstack-tox-docs/a8f0754/html/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"e2269a0c2b3e133e2148e75e6c30482ef70f4925","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-24 10:04:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8ea10b62b10144b8b26e6e0b29c2f5a9d79bc50c","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-05-02 15:46:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"68f8a0cdabc61bb4fed594b312bd6370f1a09440","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 16:39:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e44730a8d96b77c6e8eb6c3dafb229a6e6e4530a","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-03 18:32:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3ec38564e5fb6c553ffc408d75dd44d330fa8e79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:32:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b53f85ea0058b7ac8a7986acb6bd2e3814b26d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:44:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/24/655024/2/gate/test-release-openstack/1f0e3c8/ : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/24/655024/2/gate/openstack-tox-lower-constraints/28a5ea0/ : SUCCESS in 7m 17s\n- openstack-tox-pep8 http://logs.openstack.org/24/655024/2/gate/openstack-tox-pep8/548019e/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/24/655024/2/gate/openstack-tox-py27/b7df3e4/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/24/655024/2/gate/openstack-tox-py36/6fc8334/ : SUCCESS in 4m 25s\n- openstack-tox-docs http://logs.openstack.org/24/655024/2/gate/openstack-tox-docs/69f6e50/html/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"5d4d9d28e7ddbe2721892120e71cfe387c5308cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:44:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"afa85dc796edd69728f4a76f13173572749f965d","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-07-19 16:58:36.000000000","message":"Patch Set 2:\n\nthanks for merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"acf22b2cb496713eddb28b2af6785f99dcf2d043","revisions":{"d39dc804c82f6d3954dbb1c01bd5043f73f23677":{"kind":"REWORK","_number":1,"created":"2019-04-23 11:38:39.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/24/655024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mox3","ref":"refs/changes/24/655024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/24/655024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/24/655024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/24/655024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mox3 refs/changes/24/655024/1"}}},"commit":{"parents":[{"commit":"00867afff8edbc9263a42f8e1848dec2eb5b808b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/00867afff8edbc9263a42f8e1848dec2eb5b808b"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:38:21.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:38:36.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I613b5676676215c6f80d309799a2efd715894ee7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/d39dc804c82f6d3954dbb1c01bd5043f73f23677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/d39dc804c82f6d3954dbb1c01bd5043f73f23677"}]},"branch":"refs/heads/master"},"acf22b2cb496713eddb28b2af6785f99dcf2d043":{"kind":"REWORK","_number":2,"created":"2019-04-23 13:34:45.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/24/655024/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mox3","ref":"refs/changes/24/655024/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/24/655024/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/24/655024/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/24/655024/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mox3 refs/changes/24/655024/2"}}},"commit":{"parents":[{"commit":"00867afff8edbc9263a42f8e1848dec2eb5b808b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/00867afff8edbc9263a42f8e1848dec2eb5b808b"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:38:21.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 13:34:36.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I613b5676676215c6f80d309799a2efd715894ee7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/acf22b2cb496713eddb28b2af6785f99dcf2d043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/acf22b2cb496713eddb28b2af6785f99dcf2d043"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
