)]}'
{"id":"openstack%2Fmox3~669457","triplet_id":"openstack%2Fmox3~master~I09c16adfa6feeb4ebe373ee5334aa8cce5a1306d","project":"openstack/mox3","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I09c16adfa6feeb4ebe373ee5334aa8cce5a1306d","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 18:41:23.000000000","updated":"2019-07-16 21:00:01.000000000","submitted":"2019-07-16 21:00:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669457-1563310801697-9977374c","meta_rev_id":"92c35a6484ca0989b52b3b1cf92f5289cee70167","_number":669457,"virtual_id_number":669457,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":2,"date":"2019-07-16 21:00:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":2,"date":"2019-07-16 20:41:27.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":1,"date":"2019-07-16 20:48:45.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 20:48:45.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-07-16 21:00:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a6aba17e765125ac6c52e033ccd6ec93d1035a9","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 18:41:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4838e52b9055015499be0ecc98dbcacfe0bb7bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 18:49:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/57/669457/1/check/test-release-openstack/a0d0e8c/ : SUCCESS in 1m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/669457/1/check/openstack-tox-lower-constraints/105f626/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/57/669457/1/check/openstack-tox-pep8/630d068/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/57/669457/1/check/openstack-tox-py27/81e0369/ : SUCCESS in 4m 30s\n- openstack-tox-py36 http://logs.openstack.org/57/669457/1/check/openstack-tox-py36/3013a7f/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/57/669457/1/check/openstack-tox-py37/281d3b8/ : SUCCESS in 3m 23s\n- openstack-tox-docs http://logs.openstack.org/57/669457/1/check/openstack-tox-docs/ec48d81/html/ : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"8d15235abf9cc373752c6ecf1af77b36c93af54f","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-07-16 20:41:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1ddda6e8e2618719a9262608260eb1bcb03076aa","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-07-16 20:48:45.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2db988b61cef352d0a29140ce8e1cac14451dab2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 20:48:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"486c050e013f9823b737cd59121883f4656167d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 21:00:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/57/669457/1/gate/test-release-openstack/33cdd1c/ : SUCCESS in 2m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/669457/1/gate/openstack-tox-lower-constraints/7aa4e8b/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/57/669457/1/gate/openstack-tox-pep8/83fcbe2/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/57/669457/1/gate/openstack-tox-py27/8f3969f/ : SUCCESS in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/57/669457/1/gate/openstack-tox-py36/22db751/ : SUCCESS in 3m 13s\n- openstack-tox-py37 http://logs.openstack.org/57/669457/1/gate/openstack-tox-py37/ea21a4d/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/57/669457/1/gate/openstack-tox-docs/e8e5c1c/html/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"24967f95d313faf963bde5149c48dba853b13f66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 21:00:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a1e5b96f423ebd10cf1c554cf51847d0436b95f","revisions":{"0a1e5b96f423ebd10cf1c554cf51847d0436b95f":{"kind":"REWORK","_number":1,"created":"2019-07-05 18:41:23.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/57/669457/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mox3","ref":"refs/changes/57/669457/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/57/669457/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/57/669457/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/57/669457/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mox3 refs/changes/57/669457/1"}}},"commit":{"parents":[{"commit":"a176de01d73739b7084f808c35c0a574aa62f843","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/a176de01d73739b7084f808c35c0a574aa62f843"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:56:45.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 18:41:22.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I09c16adfa6feeb4ebe373ee5334aa8cce5a1306d\nStory: #2005924\nTask: #34234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/0a1e5b96f423ebd10cf1c554cf51847d0436b95f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/0a1e5b96f423ebd10cf1c554cf51847d0436b95f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
