)]}'
{"id":"openstack%2Fmox3~683471","triplet_id":"openstack%2Fmox3~stable%2Ftrain~I64c1990d098c3babec41d2b0ab2ab198c8c2012a","project":"openstack/mox3","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"I64c1990d098c3babec41d2b0ab2ab198c8c2012a","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","status":"MERGED","created":"2019-09-20 16:25:32.000000000","updated":"2019-09-23 12:13:54.000000000","submitted":"2019-09-23 12:11:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683471-1569240694789-f233d7c7","meta_rev_id":"8e0323f0f7a34d17784fd13fb33a5b354d414b92","_number":683471,"virtual_id_number":683471,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-09-23 12:11:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-20 16:38:06.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-23 11:53:48.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-23 11:53:48.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 16:38:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-23 11:53:48.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-09-23 12:11:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1df785cbea26acb73d6924f1818907b701d640cd","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-09-20 16:25:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00ca9054f8cb233553f526a99a5d9b541e5d30fb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-20 16:38:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"63a3bce4244642fd03fad72400f3f24c6973b00c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 17:20:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3558c38b6c546dd9a01e0d544286c56 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f973987dcbbe4a1286701b98f184413d : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dfd446139c424f248b022a5efb08524b : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6892edd7594844f4a251c66e283089be : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d6c79a8a70047a6bded8b9e98a49166 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39d0f31639314bc9b513beca0afcb280 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3a313acadf18abd24fc8a0cbdc48e057b2a50df1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-23 11:53:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ef8cd72aeb06557abe3656cc68abc9bd6a4059a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 11:55:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d74a72eba8e615eb348b2269a7d9c4751445325c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 12:11:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16355eb68c73484397c52fb7ec1d5d21 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca2051a2e92648199a80b6b0c116f729 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/485cf9e5fa2d4cd59411f53be96ad360 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6027ecf0d8441cfa13b72b8bc50e5c3 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8dd75c5d6bd5405e9f6b48b0afa8b573 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc674da097a4a149f9bb6391b4952c7 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"fb96e3604b70e6f5a338ca17ec7765da3282ccb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 12:11:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a301ef1fe95b79fdd2b02b844a7d43a35bf49422","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 12:13:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d5e2152607b46b08479966633280b15 : SUCCESS in 1m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e260f5061a15b08ba3386137f2c5696ce1d171ec","revisions":{"e260f5061a15b08ba3386137f2c5696ce1d171ec":{"kind":"REWORK","_number":1,"created":"2019-09-20 16:25:32.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/71/683471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mox3","ref":"refs/changes/71/683471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/71/683471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/71/683471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/71/683471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mox3 refs/changes/71/683471/1"}}},"commit":{"parents":[{"commit":"29f0ffc213c26ae2ef40416bbf2508b1d9b325ac","subject":"Update .gitreview for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/29f0ffc213c26ae2ef40416bbf2508b1d9b325ac"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-09-20 16:25:28.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-09-20 16:25:28.000000000","tz":0},"subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","message":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/train branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I64c1990d098c3babec41d2b0ab2ab198c8c2012a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/e260f5061a15b08ba3386137f2c5696ce1d171ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/e260f5061a15b08ba3386137f2c5696ce1d171ec"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
