)]}'
{"id":"openstack%2Fmox3~700679","triplet_id":"openstack%2Fmox3~master~Ibac9edb043fa082322ed2f2272cc0b1727ae9009","project":"openstack/mox3","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Ibac9edb043fa082322ed2f2272cc0b1727ae9009","subject":"Drop python 2.7 support and testing","status":"MERGED","created":"2019-12-27 07:22:00.000000000","updated":"2020-02-03 12:07:31.000000000","submitted":"2020-02-03 12:02:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700679-1580731363847-9ea6cd4b","meta_rev_id":"cb444d201991e808fe572ad7411db133ace84d37","_number":700679,"virtual_id_number":700679,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-03 12:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2020-02-03 11:48:08.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"date":"2020-02-02 11:36:54.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":1,"date":"2020-02-03 11:48:08.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-08 19:42:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-01-21 06:51:27.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-01-22 01:35:39.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-02-02 11:36:54.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-02-03 11:48:08.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-02-03 12:02:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ebf200cec5b767ca000fb32de580f60a5307ee9","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-12-27 07:22:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f28228fa9bdd9d6c8728975078814ba83449e75a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-27 07:26:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cbdb390b19104c2e942ceb7c4cafba69 : SUCCESS in 2m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/636ea387cd4a49bfa082d4e0d6414171 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37ca6d4c65a243b9aff69839d52be7db : SUCCESS in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7eb3e52aeb134f84b5fe1561e73336e3 : SUCCESS in 2m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73177506101f47d79243f053d00735e5 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48477422d3b94d27ab20702512671626 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"933a554195d90cb29c3781ff909c0a97549e31da","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-08 18:16:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c087024712eeb7893802c04ac098abe321c950f4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-08 19:29:30.000000000","message":"Patch Set 1: -Code-Review\n\nHum sorry I think we still need to keep support in oslo world while other projects are still support py2.\n\nLet\u0027s discuss ASAP about the dropping status in the whole openstack.","accounts_in_message":[],"_revision_number":1},{"id":"7eae918dacec8a965d72ba6f450e8f539150170a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-08 19:42:52.000000000","message":"Patch Set 1:\n\nSo we are in phase 2 and it\u0027s ok to drop py2 on oslo side.\n\nAnyway, we want to discuss about this during the next oslo meeting before moving further.","accounts_in_message":[],"_revision_number":1},{"id":"9cdf081468f63d9ca23dfa7450eeba330767cbf0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-15 15:16:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71098b71a21d1bdd6a4efb25b841164fc1a7234b","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-21 06:51:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"97f09c8126b20befc82e326602cb6625d1cee64d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-22 01:35:39.000000000","message":"Patch Set 1: Code-Review-1\n\nalong with review comments, you need to cleanup requirements also. Do all things mentioned in https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html#completion-criteria","accounts_in_message":[],"_revision_number":1},{"id":"9d0b568ba8bed4d56e2eafe89d51966bd1896edb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-02 11:36:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c63ed1f1f9f891c2b2e77f72d9024aee3dcde24f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 11:45:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0a2ec1f245a2452f96cb3300213a387e : SUCCESS in 2m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/89437b887e1b4cd58584398f929782a6 : SUCCESS in 2m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5374978663fb4536b3e8a248c28e2246 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb73ff87b9e64c8fa3f2e3ffc755ea34 : SUCCESS in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca529ef2a0be4349893c7a8751b9a2a2 : SUCCESS in 2m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e5fb01678614508ad98a473917e048b : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33b974a4f704434da2e1eb13f4d7bd90 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fefb463a8d304a6497897330ae9033db : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"05719d62150624642df677947596f28b71a733c4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-03 11:48:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"36e8d9480ce2e605b84406b373246b594a0d5c26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 11:48:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e1915b9cf992b1ec7a6e797f665d612094a6c15f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 12:02:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8c250f2cf94d4053a82675f10c383b20 : SUCCESS in 5m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c4d9a9b44922468180241b13798c8784 : SUCCESS in 3m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02eb6b629b37495aa045a1828af14724 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f394397351ce4b8491f85bf313118e44 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b90025383a864ba69f9787cfa4c05c8c : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba180fec982f44589e3bf414cb79dd7b : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b14eafcb40411c8ac4239ae6c9d6c7 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"06b80102b8e85ff79a4299eb6256e16b33f266af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 12:02:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dbbd74efee09bd3c32bc4f6bdf4e0d784ecb2915","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 12:07:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/570a8f1135ca45b1999302e93307746b : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"127bc84a8869723c154d25bc7ce6469448858629","revisions":{"5266178f6e89b5973995e477b5ef6e25b24af113":{"kind":"REWORK","_number":1,"created":"2019-12-27 07:22:00.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/79/700679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mox3","ref":"refs/changes/79/700679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/79/700679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/79/700679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/79/700679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mox3 refs/changes/79/700679/1"}}},"commit":{"parents":[{"commit":"6ada58bed43006fad1ebb7205f9011e1bc0921b2","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/6ada58bed43006fad1ebb7205f9011e1bc0921b2"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-12-27 07:22:54.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-12-27 07:22:54.000000000","tz":480},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Ibac9edb043fa082322ed2f2272cc0b1727ae9009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/5266178f6e89b5973995e477b5ef6e25b24af113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/5266178f6e89b5973995e477b5ef6e25b24af113"}]},"branch":"refs/heads/master"},"127bc84a8869723c154d25bc7ce6469448858629":{"kind":"REWORK","_number":2,"created":"2020-02-02 11:36:54.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/79/700679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mox3","ref":"refs/changes/79/700679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/79/700679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/79/700679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/79/700679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mox3 refs/changes/79/700679/2"}}},"commit":{"parents":[{"commit":"6ada58bed43006fad1ebb7205f9011e1bc0921b2","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/6ada58bed43006fad1ebb7205f9011e1bc0921b2"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-12-27 07:22:54.000000000","tz":480},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-02 11:36:35.000000000","tz":60},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nRemove also universal-wheel, this is py3 only now.\n\nCo-Authored-By: Andreas Jaeger \u003caj@suse.com\u003e\nChange-Id: Ibac9edb043fa082322ed2f2272cc0b1727ae9009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/127bc84a8869723c154d25bc7ce6469448858629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/127bc84a8869723c154d25bc7ce6469448858629"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
