)]}'
{"id":"openstack%2Fmox3~722705","triplet_id":"openstack%2Fmox3~master~I9889db5f710af545d11008090b455048011e38ea","project":"openstack/mox3","branch":"master","topic":"add-py38-meta","hashtags":[],"change_id":"I9889db5f710af545d11008090b455048011e38ea","subject":"Add py38 package metadata","status":"MERGED","created":"2020-04-24 13:25:07.000000000","updated":"2020-04-27 08:42:24.000000000","submitted":"2020-04-27 08:40:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722705-1587976836949-9a8ae8f1","meta_rev_id":"e414fddcd15a1c701cfc75b70db4a113e66f72ad","_number":722705,"virtual_id_number":722705,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 08:40:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-04-27 08:28:16.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-04-24 16:23:20.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-04-27 08:28:16.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-24 16:23:20.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-27 08:28:16.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-04-27 08:40:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e81e0d57b475912e4dd389f769c566a8953a946d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-24 13:25:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28afe0fb10f6c52c1b9dca91cf35c74702c1b9b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 13:38:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d5659b87c1cf41f49011c9289de9324a : SUCCESS in 4m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3308f95bb9c648468fb8ea1bbb9c2ba3 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee9376476a8d469eba891ddda08b0d5c : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57c4db386e4349e3939e8cf413994501 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/360051db3fb54be981109966bfa21a8b : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b355fd3aa5134a38bb8572f3a400e245 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"bdd7fefd949d5e0df6b8a02254ec456d51b9baf8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-24 14:35:34.000000000","message":"Patch Set 1: Code-Review+1\n\nWhat\u0027s about tox envlist?\n\nhttps://github.com/openstack/automaton/blob/master/tox.ini#L3\n\nI think we should add/move to this one here too, thoughts?","accounts_in_message":[],"_revision_number":1},{"id":"182b63d5b299531f6c21f0fde9855eec33fc54a4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-24 16:23:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e972efeca48d7ab145ca027bdab5511bb03c625e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-27 08:28:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"30c6dba1d256e21b0d0150a4561af8e18c1ed716","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 08:28:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2efff61c6c971cd61916fcc0f7f5a8f2a8afce05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 08:40:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"01f01fbbb771d569551b6183a147b787bd3c3bc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 08:40:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/31cb05c3c0fe46a298d0f5221f8b9252 : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68ee353591b54cfdaf07c2779f944ee2 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9351530b4e3d44c696e2ec56f34e05d9 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec5238cc34d0493683f3e4bef51dbd08 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/383ee2ac31304af8850a7d6661e4ee97 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d58b23bb8e24b5696a81e69986e3378 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ac13be28173d32237cd4045218fda7a7fa7793ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 08:42:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89592ccd60b64f6dbbd748058c157544 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d27eb2c38aa4fcbe5a3cff083907c76f6643690a","revisions":{"d27eb2c38aa4fcbe5a3cff083907c76f6643690a":{"kind":"REWORK","_number":1,"created":"2020-04-24 13:25:07.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/05/722705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mox3","ref":"refs/changes/05/722705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/05/722705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/05/722705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mox3 refs/changes/05/722705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mox3 refs/changes/05/722705/1"}}},"commit":{"parents":[{"commit":"7cdcc4e04fd750fc036421d701c888ff0b9ef67e","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/7cdcc4e04fd750fc036421d701c888ff0b9ef67e"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 13:22:36.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 13:22:36.000000000","tz":-300},"subject":"Add py38 package metadata","message":"Add py38 package metadata\n\nNow that we are running the Victoria tests that include a\nvoting py38, we can now add the Python 3.8 metadata to the\npackage information to reflect that support.\n\nChange-Id: I9889db5f710af545d11008090b455048011e38ea\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/d27eb2c38aa4fcbe5a3cff083907c76f6643690a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mox3/commit/d27eb2c38aa4fcbe5a3cff083907c76f6643690a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
