)]}'
{"id":"openstack%2Fmulti-arch-sig~708057","triplet_id":"openstack%2Fmulti-arch-sig~master~I3f680ea7e78ef38c32791ee70426d5de100d899a","project":"openstack/multi-arch-sig","branch":"master","topic":"story/2007257","hashtags":[],"change_id":"I3f680ea7e78ef38c32791ee70426d5de100d899a","subject":"Initial Sphinx structure","status":"MERGED","created":"2020-02-17 05:10:41.000000000","updated":"2020-02-18 17:01:56.000000000","submitted":"2020-02-18 16:55:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708057-1582044944979-5f1f21a0","meta_rev_id":"cd265755ac78fdf91bd4527c5629d294563d4cee","_number":708057,"virtual_id_number":708057,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 16:55:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":2,"date":"2020-02-18 16:47:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-18 06:44:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":1,"date":"2020-02-18 16:47:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 06:44:01.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-02-18 16:47:09.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2020-02-18 16:55:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a0d844849018a8090439b6d34d1bcb7aeadbb36","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-02-17 05:10:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0aa4a334a96802af9452ade9f53fa8d17e44a7a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 05:14:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf8b1de2c2de4a4ba0ebdd42f5f308ce : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"45e4e03dae66a22b9369954695f93a370792f9e7","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-02-17 05:21:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1fbdbad50e845c71b257fa61335e9f132ca9c62d","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-02-17 05:23:32.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cdfb44f59a21a0c290daca26c4f3d0af033076cb","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-02-17 05:23:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b11ea8ed640ed675610881617ff11aa71f555fdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 05:32:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4ebf1661d5499596c28d2b2a453972 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"185c7bee0816db54bdf216c3a7076a65ee2067fb","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2020-02-17 14:50:38.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"121b798087707871555a32a0a8c635104aa40098","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-02-18 06:21:26.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"885eba89018415a92a1856cce940879b1e18d189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 06:32:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f42126e5058d46eda01f12b32e168a59 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":3},{"id":"ec5688746ba503c8f8c302db6d8286f0c3bf8e9c","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-02-18 06:44:01.000000000","message":"Patch Set 3: Code-Review+2\n\noutput lgtm","accounts_in_message":[],"_revision_number":3},{"id":"03eb4f970908381fdf3ca7212d232b496816b701","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2020-02-18 16:47:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n:)","accounts_in_message":[],"_revision_number":3},{"id":"fe0bcfdc2a9b67d4cd6279169e8f12fb40138024","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 16:47:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"25b581c399bc27e8f67e119abeb511d7bd3b10d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 16:55:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d0ed642077c482ea0463e9f0d1d3275 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":3},{"id":"bd5f65efe401db889358f05e70d1067587d72a71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 16:55:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"20ed7c1b6c78c81d54504b0620f2216c800a8ce1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 17:01:56.000000000","message":"Patch Set 3:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b503a35a8de64fedaf7483252c46c838 : FAILURE in 3m 05s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"61d5e6c9ee32241d617afaeb8c6d02086e67e1be","revisions":{"77124094556676b5c9c680ae96c4d0f54222e784":{"kind":"REWORK","_number":1,"created":"2020-02-17 05:10:41.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/57/708057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/multi-arch-sig","ref":"refs/changes/57/708057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/57/708057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/57/708057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/57/708057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/multi-arch-sig refs/changes/57/708057/1"}}},"commit":{"parents":[{"commit":"f048f509cf206b198c5f86488831c2ce69a3aa07","subject":"Added .gitreview","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/f048f509cf206b198c5f86488831c2ce69a3aa07"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2020-02-17 05:07:27.000000000","tz":480},"committer":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2020-02-17 05:10:39.000000000","tz":480},"subject":"Initial Sphinx structure","message":"Initial Sphinx structure\n\nChange-Id: I3f680ea7e78ef38c32791ee70426d5de100d899a\nStory: 2007257\nTask: 38618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/77124094556676b5c9c680ae96c4d0f54222e784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/77124094556676b5c9c680ae96c4d0f54222e784"}]},"branch":"refs/heads/master"},"a291117d979d1235a1e520da14356056a15acb83":{"kind":"REWORK","_number":2,"created":"2020-02-17 05:23:32.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/57/708057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/multi-arch-sig","ref":"refs/changes/57/708057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/57/708057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/57/708057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/57/708057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/multi-arch-sig refs/changes/57/708057/2"}}},"commit":{"parents":[{"commit":"f048f509cf206b198c5f86488831c2ce69a3aa07","subject":"Added .gitreview","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/f048f509cf206b198c5f86488831c2ce69a3aa07"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2020-02-17 05:07:27.000000000","tz":480},"committer":{"name":"Rico Lin","email":"rico.lin.guanyu@gmail.com","date":"2020-02-17 05:23:28.000000000","tz":0},"subject":"Initial Sphinx structure","message":"Initial Sphinx structure\n\nChange-Id: I3f680ea7e78ef38c32791ee70426d5de100d899a\nStory: 2007257\nTask: 38618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/a291117d979d1235a1e520da14356056a15acb83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/a291117d979d1235a1e520da14356056a15acb83"}]},"branch":"refs/heads/master"},"61d5e6c9ee32241d617afaeb8c6d02086e67e1be":{"kind":"REWORK","_number":3,"created":"2020-02-18 06:21:26.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/57/708057/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/multi-arch-sig","ref":"refs/changes/57/708057/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/57/708057/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/57/708057/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/multi-arch-sig refs/changes/57/708057/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/multi-arch-sig refs/changes/57/708057/3"}}},"commit":{"parents":[{"commit":"f048f509cf206b198c5f86488831c2ce69a3aa07","subject":"Added .gitreview","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/f048f509cf206b198c5f86488831c2ce69a3aa07"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2020-02-17 05:07:27.000000000","tz":480},"committer":{"name":"Rico Lin","email":"rico.lin.guanyu@gmail.com","date":"2020-02-18 06:21:22.000000000","tz":0},"subject":"Initial Sphinx structure","message":"Initial Sphinx structure\n\nChange-Id: I3f680ea7e78ef38c32791ee70426d5de100d899a\nStory: 2007257\nTask: 38618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/61d5e6c9ee32241d617afaeb8c6d02086e67e1be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/multi-arch-sig/commit/61d5e6c9ee32241d617afaeb8c6d02086e67e1be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
